fix Dropdown docs & types - #1046
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 4b26119 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
|
This pull request is being automatically deployed with Vercel (learn more). 🔍 Inspect: https://vercel.com/primer/primer-components/n69jxebih |
colebemis
approved these changes
Feb 12, 2021
colebemis
left a comment
Contributor
There was a problem hiding this comment.
Great job 👍 Looks like you'll need to update a snapshot test
Comment on lines
+5
to
+7
| - Fixed Dropdown & Details types. | ||
|
|
||
| - Added useDetails behavior back to Dropdown |
Contributor
There was a problem hiding this comment.
I'd split these into two changesets
emplums
commented
Feb 16, 2021
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
When we updated the Details component, we didn't update
Dropdownwhich was relying on theDetailscomponent's behaviors. Consequently, we shipped a host of breaking changes toDropdownand removed support for some of the old props.This PR is basically a temporary fix to hold us over until we ship the new overlay system:
useDetailsDetailsand addeduseDetails. Since we are planning on doing a refresh of all our overlay related components soon, instead of adding this feature back in, I've just updated the docs and we'll take care of restoring functionality/deciding if we want to add that functionality back in when we do our overlay work. cc @smockleScreenshots
Please provide before/after screenshots for any visual changes
Merge checklist
index.d.ts) if necessaryTake a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.