Skip to content

fix(ActionList): description font size and truncation - #1253

Merged
dgreif merged 7 commits into
mainfrom
item-description-styles
May 27, 2021
Merged

fix(ActionList): description font size and truncation#1253
dgreif merged 7 commits into
mainfrom
item-description-styles

Conversation

@dgreif

@dgreif dgreif commented May 25, 2021

Copy link
Copy Markdown
Contributor

Fixing a couple small issues with ActionList.Item styles:

  • Description font size should be 12px instead of 14px
  • Inline descriptions should be truncated with ellipsis, rather than wrapping. I had to redo the flex styles a bit to make the main text area grow properly.

Also added a new story for stress testing these styles

Screenshots

image

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@dgreif
dgreif requested a review from smockle May 25, 2021 21:21
@changeset-bot

changeset-bot Bot commented May 25, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: eafa1b5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@primer/components Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@vercel

vercel Bot commented May 25, 2021

Copy link
Copy Markdown

This pull request is being automatically deployed with Vercel (learn more).
To see the status of your deployment, click below or on the icon next to each commit.

🔍 Inspect: https://vercel.com/primer/primer-components/9RfpK7wv4j2csBPA2NrjADhqDVRW
✅ Preview: https://primer-components-git-item-description-styles-primer.vercel.app

Comment thread src/ActionList/Item.tsx
@dgreif
dgreif force-pushed the item-description-styles branch from 8f4a462 to eafa1b5 Compare May 27, 2021 14:27
@dgreif
dgreif merged commit 25d88c4 into main May 27, 2021
@dgreif
dgreif deleted the item-description-styles branch May 27, 2021 14:42
@github-actions github-actions Bot mentioned this pull request May 27, 2021
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants