Skip to content

Add CODEOWNERS so the merging to master must be approved by a reviewer. - #100

Merged
CoolSpy3 merged 3 commits into
masterfrom
add-CODEOWNERS-file-to-require-review-by-reviewers
May 28, 2024
Merged

Add CODEOWNERS so the merging to master must be approved by a reviewer.#100
CoolSpy3 merged 3 commits into
masterfrom
add-CODEOWNERS-file-to-require-review-by-reviewers

Conversation

@brettle

@brettle brettle commented May 28, 2024

Copy link
Copy Markdown
Member

No description provided.

@brettle
brettle requested a review from CoolSpy3 May 28, 2024 01:17
Comment thread .github/CODEOWNERS
@CoolSpy3
CoolSpy3 merged commit b0ff5b7 into master May 28, 2024
@CoolSpy3
CoolSpy3 deleted the add-CODEOWNERS-file-to-require-review-by-reviewers branch May 28, 2024 09:00
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants