Skip to content

fix(fuselage): Option selected coercion - #2150

Merged
tassoevan merged 3 commits into
mainfrom
fix/option-select
Aug 10, 2026
Merged

tassoevan merged 3 commits into
mainfrom
fix/option-select

Conversation

@dougfabris

@dougfabris dougfabris commented Aug 10, 2026

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

Introduced here: #2049

Issue(s)

Further comments

CORE-2526

@changeset-bot

changeset-bot Bot commented Aug 10, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: de5c6b0

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@rocket.chat/fuselage Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@dougfabris
dougfabris marked this pull request as ready for review August 10, 2026 19:29
@tassoevan
tassoevan merged commit 9d13357 into main Aug 10, 2026
10 checks passed
@tassoevan
tassoevan deleted the fix/option-select branch August 10, 2026 19:47
@github-actions github-actions Bot mentioned this pull request Aug 10, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants