Skip to content

chore: Update @tanstack/config to v0.11.0 - #7795

Merged
lachlancollins merged 1 commit into
mainfrom
tanstack-config
Jul 26, 2024
Merged

lachlancollins merged 1 commit into
mainfrom
tanstack-config

Conversation

@lachlancollins

Copy link
Copy Markdown
Member

No description provided.

@nx-cloud

nx-cloud Bot commented Jul 26, 2024

Copy link
Copy Markdown

☁️ Nx Cloud Report

CI is running/has finished running commands for commit 628dcf6. As they complete they will appear below. Click to see the status, the terminal output, and the build insights.

📂 See all runs for this CI Pipeline Execution


🟥 Failed Commands
nx affected --targets=test:sherif,test:knip,test:eslint,test:lib,test:types,test:build,build --parallel=3
✅ Successfully ran 1 target

Sent with 💌 from NxCloud.

@pkg-pr-new

pkg-pr-new Bot commented Jul 26, 2024

Copy link
Copy Markdown

commit: 628dcf6

pnpm add https://pkg.pr.new/@tanstack/angular-query-devtools-experimental@7795
pnpm add https://pkg.pr.new/@tanstack/angular-query-experimental@7795
pnpm add https://pkg.pr.new/@tanstack/eslint-plugin-query@7795
pnpm add https://pkg.pr.new/@tanstack/query-async-storage-persister@7795
pnpm add https://pkg.pr.new/@tanstack/query-broadcast-client-experimental@7795
pnpm add https://pkg.pr.new/@tanstack/query-core@7795
pnpm add https://pkg.pr.new/@tanstack/query-devtools@7795
pnpm add https://pkg.pr.new/@tanstack/query-persist-client-core@7795
pnpm add https://pkg.pr.new/@tanstack/query-sync-storage-persister@7795
pnpm add https://pkg.pr.new/@tanstack/react-query@7795
pnpm add https://pkg.pr.new/@tanstack/react-query-devtools@7795
pnpm add https://pkg.pr.new/@tanstack/react-query-next-experimental@7795
pnpm add https://pkg.pr.new/@tanstack/react-query-persist-client@7795
pnpm add https://pkg.pr.new/@tanstack/solid-query@7795
pnpm add https://pkg.pr.new/@tanstack/solid-query-devtools@7795
pnpm add https://pkg.pr.new/@tanstack/solid-query-persist-client@7795
pnpm add https://pkg.pr.new/@tanstack/svelte-query@7795
pnpm add https://pkg.pr.new/@tanstack/svelte-query-devtools@7795
pnpm add https://pkg.pr.new/@tanstack/svelte-query-persist-client@7795
pnpm add https://pkg.pr.new/@tanstack/vue-query@7795
pnpm add https://pkg.pr.new/@tanstack/vue-query-devtools@7795

Open in Stackblitz

More templates

@lachlancollins
lachlancollins merged commit bd3296b into main Jul 26, 2024
@lachlancollins
lachlancollins deleted the tanstack-config branch July 26, 2024 00:56
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant