Skip to content

YARN-11657. Remove protobuf-2.5 from hadoop-yarn-api module (#6575) - #6580

Merged
steveloughran merged 1 commit into
apache:trunkfrom
steveloughran:build/YARN-11657-protobuf-2.5
Mar 5, 2024
Merged

steveloughran merged 1 commit into
apache:trunkfrom
steveloughran:build/YARN-11657-protobuf-2.5

Conversation

@steveloughran

Copy link
Copy Markdown
Contributor

The import of protobuf-java-2.5 in the hadoop-yarn-api module is downgraded from "compile" to "provided"

This removes it from share/hadoop/yarn/lib/protobuf-java-2.5.0.jar

It is still found under
share/hadoop/yarn/timelineservice/lib/protobuf-java-2.5.0.jar

For code changes:

  • Does the title or this PR starts with the corresponding JIRA issue id (e.g. 'HADOOP-17799. Your PR title ...')?
  • Object storage: have the integration tests been executed and the endpoint declared according to the connector-specific documentation?
  • If adding new dependencies to the code, are these dependencies licensed in a way that is compatible for inclusion under ASF 2.0?
  • If applicable, have you updated the LICENSE, LICENSE-binary, NOTICE-binary files?

@steveloughran

Copy link
Copy Markdown
Contributor Author

This is #6575 for trunk

)

The import of protobuf-java-2.5 in the hadoop-yarn-api module
is downgraded from "compile" to "provided"

This removes it from share/hadoop/yarn/lib/protobuf-java-2.5.0.jar

It is still found under
share/hadoop/yarn/timelineservice/lib/protobuf-java-2.5.0.jar

Change-Id: I89b00f0ddff29af4bfb3173ef7cbd8ba4563d3ef
@steveloughran
steveloughran force-pushed the build/YARN-11657-protobuf-2.5 branch from 13d1466 to 16ad42c Compare February 23, 2024 17:27
@hadoop-yetus

Copy link
Copy Markdown

💔 -1 overall

Vote Subsystem Runtime Logfile Comment
+0 🆗 reexec 12m 21s Docker mode activated.
_ Prechecks _
+1 💚 dupname 0m 0s No case conflicting files found.
+0 🆗 codespell 0m 0s codespell was not available.
+0 🆗 detsecrets 0m 0s detect-secrets was not available.
+0 🆗 xmllint 0m 0s xmllint was not available.
+0 🆗 shelldocs 0m 0s Shelldocs was not available.
+1 💚 @author 0m 0s The patch does not contain any @author tags.
-1 ❌ test4tests 0m 0s The patch doesn't appear to include any new or modified tests. Please justify why no new tests are needed for this patch. Also please list what manual steps were performed to verify this patch.
_ trunk Compile Tests _
+0 🆗 mvndep 14m 47s Maven dependency ordering for branch
+1 💚 mvninstall 32m 17s trunk passed
+1 💚 compile 19m 32s trunk passed with JDK Ubuntu-11.0.21+9-post-Ubuntu-0ubuntu120.04
+1 💚 compile 18m 43s trunk passed with JDK Private Build-1.8.0_392-8u392-ga-1~20.04-b08
+1 💚 mvnsite 23m 20s trunk passed
+1 💚 javadoc 8m 52s trunk passed with JDK Ubuntu-11.0.21+9-post-Ubuntu-0ubuntu120.04
+1 💚 javadoc 8m 12s trunk passed with JDK Private Build-1.8.0_392-8u392-ga-1~20.04-b08
+1 💚 shadedclient 49m 40s branch has no errors when building and testing our client artifacts.
_ Patch Compile Tests _
+0 🆗 mvndep 0m 32s Maven dependency ordering for patch
+1 💚 mvninstall 31m 22s the patch passed
+1 💚 compile 19m 59s the patch passed with JDK Ubuntu-11.0.21+9-post-Ubuntu-0ubuntu120.04
+1 💚 javac 19m 59s the patch passed
+1 💚 compile 19m 17s the patch passed with JDK Private Build-1.8.0_392-8u392-ga-1~20.04-b08
+1 💚 javac 19m 17s the patch passed
+1 💚 blanks 0m 0s The patch has no blanks issues.
+1 💚 mvnsite 16m 45s the patch passed
+1 💚 shellcheck 0m 0s No new issues.
+1 💚 javadoc 8m 40s the patch passed with JDK Ubuntu-11.0.21+9-post-Ubuntu-0ubuntu120.04
+1 💚 javadoc 8m 9s the patch passed with JDK Private Build-1.8.0_392-8u392-ga-1~20.04-b08
+1 💚 shadedclient 50m 40s patch has no errors when building and testing our client artifacts.
_ Other Tests _
-1 ❌ unit 756m 4s /patch-unit-root.txt root in the patch failed.
+1 💚 asflicense 1m 23s The patch does not generate ASF License warnings.
1072m 39s
Reason Tests
Failed junit tests hadoop.hdfs.protocol.TestBlockListAsLongs
hadoop.hdfs.server.datanode.TestLargeBlockReport
hadoop.hdfs.server.namenode.TestNamenodeRetryCache
hadoop.hdfs.tools.TestDFSAdmin
hadoop.hdfs.server.namenode.TestReconstructStripedBlocks
hadoop.hdfs.server.namenode.TestFSImage
Subsystem Report/Notes
Docker ClientAPI=1.44 ServerAPI=1.44 base: https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-6580/1/artifact/out/Dockerfile
GITHUB PR #6580
Optional Tests dupname asflicense codespell detsecrets compile javac javadoc mvninstall mvnsite unit shadedclient xmllint shellcheck shelldocs
uname Linux 0184bc318077 5.15.0-94-generic #104-Ubuntu SMP Tue Jan 9 15:25:40 UTC 2024 x86_64 x86_64 x86_64 GNU/Linux
Build tool maven
Personality dev-support/bin/hadoop.sh
git revision trunk / 16ad42c
Default Java Private Build-1.8.0_392-8u392-ga-1~20.04-b08
Multi-JDK versions /usr/lib/jvm/java-11-openjdk-amd64:Ubuntu-11.0.21+9-post-Ubuntu-0ubuntu120.04 /usr/lib/jvm/java-8-openjdk-amd64:Private Build-1.8.0_392-8u392-ga-1~20.04-b08
Test Results https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-6580/1/testReport/
Max. process+thread count 3843 (vs. ulimit of 5500)
modules C: hadoop-yarn-project/hadoop-yarn/hadoop-yarn-api hadoop-yarn-project/hadoop-yarn/hadoop-yarn-server/hadoop-yarn-server-applicationhistoryservice hadoop-yarn-project/hadoop-yarn/hadoop-yarn-server/hadoop-yarn-server-nodemanager . U: .
Console output https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-6580/1/console
versions git=2.25.1 maven=3.6.3 shellcheck=0.7.0
Powered by Apache Yetus 0.14.0 https://yetus.apache.org

This message was automatically generated.

@hadoop-yetus

Copy link
Copy Markdown

💔 -1 overall

Vote Subsystem Runtime Logfile Comment
+0 🆗 reexec 0m 34s Docker mode activated.
_ Prechecks _
+1 💚 dupname 0m 0s No case conflicting files found.
+0 🆗 codespell 0m 0s codespell was not available.
+0 🆗 detsecrets 0m 0s detect-secrets was not available.
+0 🆗 xmllint 0m 0s xmllint was not available.
+0 🆗 shelldocs 0m 0s Shelldocs was not available.
+1 💚 @author 0m 0s The patch does not contain any @author tags.
-1 ❌ test4tests 0m 0s The patch doesn't appear to include any new or modified tests. Please justify why no new tests are needed for this patch. Also please list what manual steps were performed to verify this patch.
_ trunk Compile Tests _
+0 🆗 mvndep 14m 42s Maven dependency ordering for branch
+1 💚 mvninstall 32m 4s trunk passed
+1 💚 compile 19m 17s trunk passed with JDK Ubuntu-11.0.21+9-post-Ubuntu-0ubuntu120.04
+1 💚 compile 18m 22s trunk passed with JDK Private Build-1.8.0_392-8u392-ga-1~20.04-b08
+1 💚 mvnsite 23m 10s trunk passed
+1 💚 javadoc 8m 47s trunk passed with JDK Ubuntu-11.0.21+9-post-Ubuntu-0ubuntu120.04
+1 💚 javadoc 8m 9s trunk passed with JDK Private Build-1.8.0_392-8u392-ga-1~20.04-b08
+1 💚 shadedclient 49m 38s branch has no errors when building and testing our client artifacts.
_ Patch Compile Tests _
+0 🆗 mvndep 0m 32s Maven dependency ordering for patch
+1 💚 mvninstall 30m 56s the patch passed
+1 💚 compile 19m 46s the patch passed with JDK Ubuntu-11.0.21+9-post-Ubuntu-0ubuntu120.04
+1 💚 javac 19m 46s the patch passed
+1 💚 compile 19m 0s the patch passed with JDK Private Build-1.8.0_392-8u392-ga-1~20.04-b08
+1 💚 javac 19m 0s the patch passed
+1 💚 blanks 0m 0s The patch has no blanks issues.
+1 💚 mvnsite 17m 8s the patch passed
+1 💚 shellcheck 0m 0s No new issues.
+1 💚 javadoc 8m 40s the patch passed with JDK Ubuntu-11.0.21+9-post-Ubuntu-0ubuntu120.04
+1 💚 javadoc 8m 12s the patch passed with JDK Private Build-1.8.0_392-8u392-ga-1~20.04-b08
+1 💚 shadedclient 50m 48s patch has no errors when building and testing our client artifacts.
_ Other Tests _
-1 ❌ unit 769m 45s /patch-unit-root.txt root in the patch failed.
+1 💚 asflicense 1m 37s The patch does not generate ASF License warnings.
1073m 44s
Reason Tests
Failed junit tests hadoop.hdfs.TestErasureCodingPolicyWithSnapshotWithRandomECPolicy
hadoop.hdfs.protocol.TestBlockListAsLongs
hadoop.hdfs.server.datanode.TestLargeBlockReport
hadoop.hdfs.server.namenode.TestCacheDirectivesWithViewDFS
hadoop.hdfs.server.namenode.TestCacheDirectives
hadoop.hdfs.tools.TestDFSAdmin
hadoop.hdfs.server.namenode.TestReconstructStripedBlocks
hadoop.hdfs.server.namenode.TestNameNodeMXBean
Subsystem Report/Notes
Docker ClientAPI=1.44 ServerAPI=1.44 base: https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-6580/2/artifact/out/Dockerfile
GITHUB PR #6580
Optional Tests dupname asflicense codespell detsecrets compile javac javadoc mvninstall mvnsite unit shadedclient xmllint shellcheck shelldocs
uname Linux fa7dfa5768b6 5.15.0-94-generic #104-Ubuntu SMP Tue Jan 9 15:25:40 UTC 2024 x86_64 x86_64 x86_64 GNU/Linux
Build tool maven
Personality dev-support/bin/hadoop.sh
git revision trunk / 16ad42c
Default Java Private Build-1.8.0_392-8u392-ga-1~20.04-b08
Multi-JDK versions /usr/lib/jvm/java-11-openjdk-amd64:Ubuntu-11.0.21+9-post-Ubuntu-0ubuntu120.04 /usr/lib/jvm/java-8-openjdk-amd64:Private Build-1.8.0_392-8u392-ga-1~20.04-b08
Test Results https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-6580/2/testReport/
Max. process+thread count 3750 (vs. ulimit of 5500)
modules C: hadoop-yarn-project/hadoop-yarn/hadoop-yarn-api hadoop-yarn-project/hadoop-yarn/hadoop-yarn-server/hadoop-yarn-server-applicationhistoryservice hadoop-yarn-project/hadoop-yarn/hadoop-yarn-server/hadoop-yarn-server-nodemanager . U: .
Console output https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-6580/2/console
versions git=2.25.1 maven=3.6.3 shellcheck=0.7.0
Powered by Apache Yetus 0.14.0 https://yetus.apache.org

This message was automatically generated.

@dineshchitlangia dineshchitlangia left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

+1 LGTM, thanks Steve.

@steveloughran

Copy link
Copy Markdown
Contributor Author

Can I hold back on this and go for the "cut hbase 1 timeline service" I think stuff may fall out from that automatically. Then we can revisit this to see what to do

@steveloughran
steveloughran merged commit dae871e into apache:trunk Mar 5, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants