You signed in with another tab or window. Reload to refresh your session.You signed out in another tab or window. Reload to refresh your session.You switched accounts on another tab or window. Reload to refresh your session.Dismiss alert
As mentioned in HDFS-16739, in order to satisfy the storage policy, the length of targetIndices will be more than the actual number of blocks that need to be reconstructed.
HDFS-16776 may cause the reconstruction task to never succeed to satisfy the storage policy (unless the mover is manually triggered to make the block satisfy the storage policy).
The patch doesn't appear to include any new or modified tests. Please justify why no new tests are needed for this patch. Also please list what manual steps were performed to verify this patch.
_ trunk Compile Tests _
+1 💚
mvninstall
46m 32s
trunk passed
+1 💚
compile
1m 21s
trunk passed with JDK Ubuntu-11.0.23+9-post-Ubuntu-1ubuntu120.04.2
+1 💚
compile
1m 19s
trunk passed with JDK Private Build-1.8.0_412-8u412-ga-1~20.04.1-b08
+1 💚
checkstyle
1m 12s
trunk passed
+1 💚
mvnsite
1m 26s
trunk passed
+1 💚
javadoc
1m 8s
trunk passed with JDK Ubuntu-11.0.23+9-post-Ubuntu-1ubuntu120.04.2
+1 💚
javadoc
1m 41s
trunk passed with JDK Private Build-1.8.0_412-8u412-ga-1~20.04.1-b08
+1 💚
spotbugs
3m 19s
trunk passed
+1 💚
shadedclient
35m 49s
branch has no errors when building and testing our client artifacts.
_ Patch Compile Tests _
+1 💚
mvninstall
1m 11s
the patch passed
+1 💚
compile
1m 13s
the patch passed with JDK Ubuntu-11.0.23+9-post-Ubuntu-1ubuntu120.04.2
+1 💚
javac
1m 13s
the patch passed
+1 💚
compile
1m 7s
the patch passed with JDK Private Build-1.8.0_412-8u412-ga-1~20.04.1-b08
The patch doesn't appear to include any new or modified tests. Please justify why no new tests are needed for this patch. Also please list what manual steps were performed to verify this patch.
_ trunk Compile Tests _
+1 💚
mvninstall
45m 3s
trunk passed
+1 💚
compile
1m 22s
trunk passed with JDK Ubuntu-11.0.23+9-post-Ubuntu-1ubuntu120.04.2
+1 💚
compile
1m 16s
trunk passed with JDK Private Build-1.8.0_412-8u412-ga-1~20.04.1-b08
+1 💚
checkstyle
1m 12s
trunk passed
+1 💚
mvnsite
1m 23s
trunk passed
+1 💚
javadoc
1m 9s
trunk passed with JDK Ubuntu-11.0.23+9-post-Ubuntu-1ubuntu120.04.2
+1 💚
javadoc
1m 49s
trunk passed with JDK Private Build-1.8.0_412-8u412-ga-1~20.04.1-b08
+1 💚
spotbugs
3m 21s
trunk passed
+1 💚
shadedclient
35m 46s
branch has no errors when building and testing our client artifacts.
_ Patch Compile Tests _
+1 💚
mvninstall
1m 12s
the patch passed
+1 💚
compile
1m 16s
the patch passed with JDK Ubuntu-11.0.23+9-post-Ubuntu-1ubuntu120.04.2
+1 💚
javac
1m 16s
the patch passed
+1 💚
compile
1m 9s
the patch passed with JDK Private Build-1.8.0_412-8u412-ga-1~20.04.1-b08
+1 💚
javac
1m 9s
the patch passed
+1 💚
blanks
0m 0s
The patch has no blanks issues.
+1 💚
checkstyle
0m 56s
the patch passed
+1 💚
mvnsite
1m 14s
the patch passed
+1 💚
javadoc
0m 54s
the patch passed with JDK Ubuntu-11.0.23+9-post-Ubuntu-1ubuntu120.04.2
+1 💚
javadoc
1m 37s
the patch passed with JDK Private Build-1.8.0_412-8u412-ga-1~20.04.1-b08
+1 💚
spotbugs
3m 14s
the patch passed
+1 💚
shadedclient
35m 24s
patch has no errors when building and testing our client artifacts.
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
JIRA: HDFS-16739.
As mentioned in HDFS-16739, in order to satisfy the storage policy, the length of targetIndices will be more than the actual number of blocks that need to be reconstructed.
HDFS-16776 may cause the reconstruction task to never succeed to satisfy the storage policy (unless the mover is manually triggered to make the block satisfy the storage policy).