Skip to content

HADOOP-19502. Add RegexpSingleline module to checkstyle.xml. - #7505

Merged
slfan1989 merged 1 commit into
apache:trunkfrom
hfutatzhanghb:HADOOP-19502
Mar 25, 2025
Merged

slfan1989 merged 1 commit into
apache:trunkfrom
hfutatzhanghb:HADOOP-19502

Conversation

@hfutatzhanghb

@hfutatzhanghb hfutatzhanghb commented Mar 14, 2025

Copy link
Copy Markdown
Member

Description of PR

JIRA: HADOOP-19502. Add RegexpSingleline module to checkstyle.xml.

Add RegexpSingleline module to checkstyle.xml.
In my local, checkstyle tools use checkstyle.xml can not detect blanks.

@github-actions github-actions Bot added the trunk label Mar 14, 2025
@hadoop-yetus

Copy link
Copy Markdown

💔 -1 overall

Vote Subsystem Runtime Logfile Comment
+0 🆗 reexec 0m 31s Docker mode activated.
_ Prechecks _
+1 💚 dupname 0m 0s No case conflicting files found.
+0 🆗 codespell 0m 0s codespell was not available.
+0 🆗 detsecrets 0m 0s detect-secrets was not available.
+0 🆗 xmllint 0m 0s xmllint was not available.
+1 💚 @author 0m 0s The patch does not contain any @author tags.
-1 ❌ test4tests 0m 0s The patch doesn't appear to include any new or modified tests. Please justify why no new tests are needed for this patch. Also please list what manual steps were performed to verify this patch.
_ trunk Compile Tests _
+1 💚 mvninstall 36m 5s trunk passed
+1 💚 compile 0m 24s trunk passed with JDK Ubuntu-11.0.26+4-post-Ubuntu-1ubuntu120.04
+1 💚 compile 0m 24s trunk passed with JDK Private Build-1.8.0_442-8u442-b06us1-0ubuntu120.04-b06
+1 💚 mvnsite 0m 24s trunk passed
+1 💚 javadoc 0m 33s trunk passed with JDK Ubuntu-11.0.26+4-post-Ubuntu-1ubuntu120.04
+1 💚 javadoc 0m 24s trunk passed with JDK Private Build-1.8.0_442-8u442-b06us1-0ubuntu120.04-b06
+1 💚 shadedclient 71m 24s branch has no errors when building and testing our client artifacts.
_ Patch Compile Tests _
+1 💚 mvninstall 0m 16s the patch passed
+1 💚 compile 0m 14s the patch passed with JDK Ubuntu-11.0.26+4-post-Ubuntu-1ubuntu120.04
+1 💚 javac 0m 14s the patch passed
+1 💚 compile 0m 14s the patch passed with JDK Private Build-1.8.0_442-8u442-b06us1-0ubuntu120.04-b06
+1 💚 javac 0m 14s the patch passed
+1 💚 blanks 0m 0s The patch has no blanks issues.
+1 💚 mvnsite 0m 14s the patch passed
+1 💚 javadoc 0m 14s the patch passed with JDK Ubuntu-11.0.26+4-post-Ubuntu-1ubuntu120.04
+1 💚 javadoc 0m 15s the patch passed with JDK Private Build-1.8.0_442-8u442-b06us1-0ubuntu120.04-b06
+1 💚 shadedclient 34m 26s patch has no errors when building and testing our client artifacts.
_ Other Tests _
+1 💚 unit 0m 19s hadoop-build-tools in the patch passed.
+1 💚 asflicense 0m 38s The patch does not generate ASF License warnings.
109m 44s
Subsystem Report/Notes
Docker ClientAPI=1.48 ServerAPI=1.48 base: https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-7505/1/artifact/out/Dockerfile
GITHUB PR #7505
Optional Tests dupname asflicense compile javac javadoc mvninstall mvnsite unit shadedclient codespell detsecrets xmllint
uname Linux 243e431c62db 5.15.0-134-generic #145-Ubuntu SMP Wed Feb 12 20:08:39 UTC 2025 x86_64 x86_64 x86_64 GNU/Linux
Build tool maven
Personality dev-support/bin/hadoop.sh
git revision trunk / 91ab935
Default Java Private Build-1.8.0_442-8u442-b06us1-0ubuntu120.04-b06
Multi-JDK versions /usr/lib/jvm/java-11-openjdk-amd64:Ubuntu-11.0.26+4-post-Ubuntu-1ubuntu120.04 /usr/lib/jvm/java-8-openjdk-amd64:Private Build-1.8.0_442-8u442-b06us1-0ubuntu120.04-b06
Test Results https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-7505/1/testReport/
Max. process+thread count 623 (vs. ulimit of 5500)
modules C: hadoop-build-tools U: hadoop-build-tools
Console output https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-7505/1/console
versions git=2.25.1 maven=3.6.3
Powered by Apache Yetus 0.14.0 https://yetus.apache.org

This message was automatically generated.

@hfutatzhanghb

Copy link
Copy Markdown
Member Author

Hi, @slfan1989 . Could you please help review this PR when have free time? Thanks ahead!

@slfan1989

Copy link
Copy Markdown
Contributor

@hfutatzhanghb Thanks for the contribution! LGTM.

@slfan1989

Copy link
Copy Markdown
Contributor

@steveloughran Could you please review this pr? Thank you very much!

@slfan1989

Copy link
Copy Markdown
Contributor

@hfutatzhanghb If there are no other comments, I will merge this pr tomorrow.

@hfutatzhanghb

Copy link
Copy Markdown
Member Author

@hfutatzhanghb If there are no other comments, I will merge this pr tomorrow.

@slfan1989 Thanks a lot!!!

@slfan1989
slfan1989 merged commit 488376c into apache:trunk Mar 25, 2025
@slfan1989

Copy link
Copy Markdown
Contributor

@hfutatzhanghb Thanks for the contribution!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants