Skip to content

HADOOP-19965. GHA update build status runs in forked repos - #8702

Merged
steveloughran merged 1 commit into
apache:trunkfrom
steveloughran:build/HADOOP-19965-gha-spam
Aug 26, 2026
Merged

steveloughran merged 1 commit into
apache:trunkfrom
steveloughran:build/HADOOP-19965-gha-spam

Conversation

@steveloughran

Copy link
Copy Markdown
Contributor

Description of PR

shuts down build status run on forks

How was this patch tested?

pushed it up as trunk to my own repo to see what happens

For code changes:

  • Does the title of this PR start with the corresponding JIRA issue id (e.g. 'HADOOP-17799. Your PR title ...')?
  • Object storage: Have the integration tests been executed and the endpoint
    declared according to the connector-specific documentation? Note: Automated CI
    testing doesn't cover all cases so manual testing with cloud storage is still
    required.
  • If adding new dependencies to the code, are these dependencies licensed in a way that is compatible for inclusion under ASF 2.0?
  • If applicable, have you updated the LICENSE, LICENSE-binary, NOTICE-binary files?

AI Tooling

If an AI tool was used:

@hadoop-yetus

Copy link
Copy Markdown

💔 -1 overall

Vote Subsystem Runtime Logfile Comment
+0 🆗 reexec 18m 40s Docker mode activated.
_ Prechecks _
+1 💚 dupname 0m 0s No case conflicting files found.
+0 🆗 codespell 0m 0s codespell was not available.
+0 🆗 detsecrets 0m 0s detect-secrets was not available.
+0 🆗 yamllint 0m 0s yamllint was not available.
+1 💚 @author 0m 0s The patch does not contain any @author tags.
_ trunk Compile Tests _
+1 💚 shadedclient 37m 20s branch has no errors when building and testing our client artifacts.
_ Patch Compile Tests _
+1 💚 blanks 0m 0s The patch has no blanks issues.
+1 💚 shadedclient 33m 31s patch has no errors when building and testing our client artifacts.
_ Other Tests _
-1 ❌ asflicense 0m 37s /results-asflicense.txt The patch generated 1 ASF License warnings.
91m 52s
Subsystem Report/Notes
Docker ClientAPI=1.55 ServerAPI=1.55 base: https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-8702/1/artifact/out/Dockerfile
GITHUB PR #8702
Optional Tests dupname asflicense codespell detsecrets yamllint
uname Linux c0baf40b72ea 5.15.0-181-generic #191-Ubuntu SMP Fri May 22 19:09:02 UTC 2026 x86_64 x86_64 x86_64 GNU/Linux
Build tool maven
Personality dev-support/bin/hadoop.sh
git revision trunk / db117bc
Max. process+thread count 572 (vs. ulimit of 10000)
modules C: . U: .
Console output https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-8702/1/console
versions git=2.43.0 maven=3.9.15
Powered by Apache Yetus 0.14.1 https://yetus.apache.org

This message was automatically generated.

@ajfabbri ajfabbri left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM. We can always modify this locally when we need to test fork of fork workflows (i.e. when we're hacking on github actions CI)

@steveloughran
steveloughran merged commit 26b0b5a into apache:trunk Aug 26, 2026
5 of 8 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants