Skip to content

Don't create a PR if all changesets are empty - #206

Merged
Andarist merged 3 commits into
changesets:mainfrom
glasser:glasser/no-empty-prs
Aug 18, 2022
Merged

Don't create a PR if all changesets are empty#206
Andarist merged 3 commits into
changesets:mainfrom
glasser:glasser/no-empty-prs

Conversation

@glasser

Copy link
Copy Markdown
Contributor

Fixes#205.

Note that this maintains the previous behavior that if there are any
changesets (even empty ones), the action does not publish packages.

Fixeschangesets#205.
Note that this maintains the previous behavior that if there are any
changesets (even empty ones), the action does not publish packages.
@changeset-bot

changeset-botBot commented Aug 18, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 452c4ea

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@changesets/actionPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@glasser

Copy link
Copy Markdown
ContributorAuthor

I didn't add a test because index.ts has no tests and I was a bit intimidated by the idea of writing the first one.

Comment thread.changeset/three-needles-protect.md Outdated
@Andarist
Andarist merged commit 8c3f5f5 into changesets:mainAug 18, 2022
@github-actionsgithub-actionsBot mentioned this pull request Aug 18, 2022
cometkim pushed a commit to cometkim/yarn-changeset-action that referenced this pull request Apr 11, 2023
…#206)
* Don't create a PR if all changesets are empty
Fixeschangesets#205.
Note that this maintains the previous behavior that if there are any
changesets (even empty ones), the action does not publish packages.
* Create three-needles-protect.md
* Update .changeset/three-needles-protect.md
Co-authored-by: Mateusz Burzyński <mateuszburzynski@gmail.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

If all changesets are empty, the action still creates a PR

2 participants

@glasser@Andarist