Skip to content

Feature: support dynamic mount for ossfs - #3432

Merged
fluid-e2e-bot[bot] merged 16 commits into
fluid-cloudnative:masterfrom
wangshli:dynamic_mount
May 9, 2024
Merged

fluid-e2e-bot[bot] merged 16 commits into
fluid-cloudnative:masterfrom
wangshli:dynamic_mount

Conversation

@wangshli

@wangshli wangshli commented Sep 1, 2023

Copy link
Copy Markdown
Contributor

Ⅰ. Describe what this PR does

support dynamic mount and add example based on ossfs.

  • thinruntime controller adds watch to dataset
  • support updating FUSE config when changed

[IMPORTANT] BREAKING CHANGE: In this PR, we changed the mounting path of that FUSE Configmap from /etc/fluid/config.json to /etc/fluid/config/config.json. This may require users to re-build their FUSE image when using new version of Fluid since v1.0.1

Ⅱ. Does this pull request fix one issue?

fixes #XXXX

Ⅲ. List the added test cases (unit test/integration test) if any, please explain if no tests are needed.

Ⅳ. Describe how to verify it

  • create dataset with thinruntime use ossfs-profile thinruntime profile
  • create app pod to mount above dataset
  • modify dataset mounts
  • see the corresponding changes in the app pod

Ⅴ. Special notes for reviews

Compatibility: change thinruntime fuse configmap mount location from /etc/fluid/config.json to /etc/fluid/config/config.json
Compatibility: change thinruntime runtimeset configmap mount location from /etc/fluid/runtime.json to /etc/fluid/runtime/runtime.json. runtimeset configmap is used for service discovery for a distributed file system. If you are not using runtimeset configmap, this should have no effect.

@fluid-e2e-bot

fluid-e2e-bot Bot commented Sep 1, 2023

Copy link
Copy Markdown

Hi @wangshli. Thanks for your PR.

I'm waiting for a fluid-cloudnative member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Comment thread charts/thin/templates/fuse/daemonset.yaml
@sonarqubecloud

sonarqubecloud Bot commented Sep 4, 2023

Copy link
Copy Markdown

Kudos, SonarCloud Quality Gate passed!    Quality Gate passed

Bug A 0 Bugs
Vulnerability A 0 Vulnerabilities
Security Hotspot E 1 Security Hotspot
Code Smell A 2 Code Smells

No Coverage information No Coverage information
0.0% 0.0% Duplication

@wangshli
wangshli force-pushed the dynamic_mount branch 2 times, most recently from 9ba0628 to 7b1a204 Compare October 8, 2023 02:37
@wangshli
wangshli marked this pull request as ready for review October 8, 2023 02:41
@codecov

codecov Bot commented Oct 8, 2023

Copy link
Copy Markdown

Codecov Report

Attention: Patch coverage is 25.89286% with 83 lines in your changes are missing coverage. Please review.

Project coverage is 64.18%. Comparing base (6d0ba26) to head (a91023d).
Report is 37 commits behind head on master.

❗ Current head a91023d differs from pull request most recent head 1c5841d. Consider uploading reports for the commit 1c5841d to get more accurate results

Files Patch % Lines
pkg/ddc/thin/ufs.go 29.85% 43 Missing and 4 partials ⚠️
pkg/ctrl/watch/dataset.go 0.00% 19 Missing ⚠️
pkg/ctrl/watch/manager.go 0.00% 14 Missing ⚠️
pkg/ddc/thin/operations/base.go 50.00% 1 Missing and 2 partials ⚠️
Additional details and impacted files
@@            Coverage Diff             @@
##           master    #3432      +/-   ##
==========================================
- Coverage   64.33%   64.18%   -0.15%     
==========================================
  Files         477      478       +1     
  Lines       28444    28547     +103     
==========================================
+ Hits        18300    18324      +24     
- Misses       7968     8043      +75     
- Partials     2176     2180       +4     

☔ View full report in Codecov by Sentry.
📢 Have feedback on the report? Share it here.

Comment thread addons/dynamic-mount/ossfs/docker/umount.sh Outdated
Comment thread pkg/ddc/thin/ufs.go
Comment thread pkg/ddc/thin/ufs.go
Comment thread addons/dynamic-mount/ossfs/docker/mount_and_umount.py Outdated
@sonarqubecloud

Copy link
Copy Markdown

Kudos, SonarCloud Quality Gate passed!    Quality Gate passed

Bug A 0 Bugs
Vulnerability A 0 Vulnerabilities
Security Hotspot E 1 Security Hotspot
Code Smell A 5 Code Smells

No Coverage information No Coverage information
0.0% 0.0% Duplication

Comment thread addons/dynamic-mount/ossfs/docker/inotify.sh
wangshli added 9 commits May 7, 2024 20:28
Signed-off-by: wangshulin <wangshulin@smail.nju.edu.cn>
Signed-off-by: wangshulin <wangshulin@smail.nju.edu.cn>
Signed-off-by: wangshulin <wangshulin@smail.nju.edu.cn>
Signed-off-by: wangshulin <wangshulin@smail.nju.edu.cn>
Signed-off-by: wangshulin <wangshulin@smail.nju.edu.cn>
Signed-off-by: wangshulin <wangshulin@smail.nju.edu.cn>
Signed-off-by: wangshulin <wangshulin@smail.nju.edu.cn>
Signed-off-by: wangshulin <wangshulin@smail.nju.edu.cn>
Signed-off-by: wangshulin <wangshulin@smail.nju.edu.cn>
Signed-off-by: trafalgarzzz <trafalgarz@outlook.com>
Signed-off-by: trafalgarzzz <trafalgarz@outlook.com>
Signed-off-by: trafalgarzzz <trafalgarz@outlook.com>
…c/fluid/config/config.json`

Signed-off-by: trafalgarzzz <trafalgarz@outlook.com>
Signed-off-by: trafalgarzzz <trafalgarz@outlook.com>
…concilerWithDataset`

Signed-off-by: trafalgarzzz <trafalgarz@outlook.com>
Comment thread pkg/controllers/v1alpha1/thinruntime/thinruntime_controller.go
Comment thread addons/dynamic-mount/ossfs/docker/build.sh
Comment thread addons/dynamic-mount/ossfs/docker/umount.sh Outdated
Comment thread api/v1alpha1/zz_generated.deepcopy.go
Signed-off-by: trafalgarzzz <trafalgarz@outlook.com>
@sonarqubecloud

sonarqubecloud Bot commented May 9, 2024

Copy link
Copy Markdown

Quality Gate Passed Quality Gate passed

Issues
8 New issues
0 Accepted issues

Measures
0 Security Hotspots
No data about Coverage
16.5% Duplication on New Code

See analysis details on SonarCloud

@cheyang cheyang left a comment

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@cheyang
cheyang requested review from TrafalgarZZZ and xliuqq and removed request for TrafalgarZZZ and xliuqq May 9, 2024 09:07
@fluid-e2e-bot

fluid-e2e-bot Bot commented May 9, 2024

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: cheyang, TrafalgarZZZ

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [TrafalgarZZZ,cheyang]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants