Skip to content

Fix v8 migration guide for Next.js to align with wizard - #12195

Merged
lforst merged 1 commit into
masterfrom
lforst-fix-nextjs-migration-guide
Jan 7, 2025
Merged

Fix v8 migration guide for Next.js to align with wizard#12195
lforst merged 1 commit into
masterfrom
lforst-fix-nextjs-migration-guide

Conversation

@lforst

Copy link
Copy Markdown
Contributor

DESCRIBE YOUR PR

getsentry/sentry-javascript#13103 (comment) raised that our migration guide is a bit outdated/wrong/could be improved. This PR fixes it.

IS YOUR CHANGE URGENT?

Help us prioritize incoming PRs by letting us know when the change needs to go live.

  • Urgent deadline (GA date, etc.):
  • Other deadline:
  • None: Not urgent, can wait up to 1 week+

@vercel

vercelBot commented Dec 20, 2024

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

NameStatusPreviewCommentsUpdated (UTC)
sentry-docs✅ Ready (Inspect)Visit Preview💬 Add feedbackDec 20, 2024 11:58am
2 Skipped Deployments
NameStatusPreviewCommentsUpdated (UTC)
changelog⬜️ Ignored (Inspect)Visit PreviewDec 20, 2024 11:58am
develop-docs⬜️ Ignored (Inspect)Visit PreviewDec 20, 2024 11:58am

@codecov

codecovBot commented Dec 20, 2024

Copy link
Copy Markdown

Bundle Report

Changes will decrease total bundle size by 15 bytes (-0.0%) ⬇️. This is within the configured threshold ✅

Detailed changes
Bundle nameSizeChange
sentry-docs-server-cjs10.36MB9 bytes (-0.0%) ⬇️
sentry-docs-client-array-push9.28MB6 bytes (-0.0%) ⬇️

@coolguyzonecoolguyzone left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks for the update!

@lforst
lforst merged commit 903a8f7 into masterJan 7, 2025
@lforst
lforst deleted the lforst-fix-nextjs-migration-guide branch January 7, 2025 10:01
@github-actionsgithub-actionsBot locked and limited conversation to collaborators Jan 23, 2025
Sign up for freeto subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@lforst@coolguyzone