Skip to content

feat(develop): Add outgoing_request context - #13204

Closed
mydea wants to merge 7 commits into
masterfrom
fn/add-outgoingRequestContext
Closed

feat(develop): Add outgoing_request context#13204
mydea wants to merge 7 commits into
masterfrom
fn/add-outgoingRequestContext

Conversation

@mydea

@mydeamydea commented Apr 2, 2025

Copy link
Copy Markdown
Member

This should replace event.request on the client completely.

Draft on top of #13203

@mydeamydea self-assigned this Apr 2, 2025
@vercel

vercelBot commented Apr 2, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

NameStatusPreviewCommentsUpdated (UTC)
develop-docs✅ Ready (Inspect)Visit Preview💬 Add feedbackApr 2, 2025 2:41pm
2 Skipped Deployments
NameStatusPreviewCommentsUpdated (UTC)
changelog⬜️ Ignored (Inspect)Visit PreviewApr 2, 2025 2:41pm
sentry-docs⬜️ Ignored (Inspect)Visit PreviewApr 2, 2025 2:41pm

{
"contexts": {
"outgoing_request": {
"url": "https://sentry.io/api/0/projects/sentry/sentry/issues/?page=1",

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Ah, I based this on https://develop.sentry.dev/sdk/data-model/event-payloads/request/#attributes:

url Optional. The URL of the request if available. The query string can be
declared either as part of the url, or separately in query_string.

We can split it out for sure!

Comment threaddevelop-docs/sdk/data-model/event-payloads/contexts.mdx Outdated
Comment threaddevelop-docs/sdk/data-model/event-payloads/contexts.mdx Outdated
@bruno-garcia

bruno-garcia commented Apr 2, 2025

Copy link
Copy Markdown
Member

This should replace event.request on the client completely.

if the product will have UX based in the new outgoing request field (which makes sense) could we have something in Relay that moves event.request to the new field when we know it's a client event? the heuristics to be defined (sdk name? other fields etc)

this way older sdks can have the benefits added by the new field

@mydea

mydea commented Apr 2, 2025

Copy link
Copy Markdown
MemberAuthor

This should replace event.request on the client completely.

if the product will have UX based in the new outgoing request field (which makes sense) could we have something in Relay that moves event.request to the new field when we know it's a client event? the heuristics to be defined (sdk name? other fields etc)

this way older sdks can have the benefits added by the new field

The idea is that today we have this overloaded thing - you can't even really set this outgoing_request information as of now for server SDKs, because event.request is already used for the incoming request information. So I do not think it will really be possible to extract this, as we can't easily differentiate what kind of data is stored in event.request 🤔

@mydea
mydeaforce-pushed the fn/add-outgoingRequestContext branch from 2ede4a0 to 8d37afaCompareApril 2, 2025 14:35
@cleptriccleptric self-assigned this Apr 3, 2025
Base automatically changed from fn/page-context to masterMay 14, 2025 14:33
@mydeamydea closed this Apr 13, 2026
@github-actionsgithub-actionsBot locked and limited conversation to collaborators Apr 29, 2026
Sign up for freeto subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@mydea@bruno-garcia@cleptric