Skip to content

CODEOWNERS file syntax documentation edited to make it explicitly clear that a pattern can only be used once - #24211

Merged
cmwilson21 merged 6 commits into
github:mainfrom
OlayinkaAtobiloye:codeowners
Mar 2, 2023
Merged

CODEOWNERS file syntax documentation edited to make it explicitly clear that a pattern can only be used once#24211
cmwilson21 merged 6 commits into
github:mainfrom
OlayinkaAtobiloye:codeowners

Conversation

@OlayinkaAtobiloye

@OlayinkaAtobiloyeOlayinkaAtobiloye commented Feb 28, 2023

Copy link
Copy Markdown
Contributor

Why:

Closes#17058

What's being changed (if available, include any code snippets, screenshots, or gifs):

Check off the following:

  • I have reviewed my changes in staging (look for the "Automatically generated comment" and click the links in the "Preview" column to view your latest changes).
  • For content changes, I have completed the self-review checklist.

@github-actionsgithub-actionsBot added the triage Do not begin working on this issue until triaged by the team label Feb 28, 2023
@OlayinkaAtobiloye

Copy link
Copy Markdown
ContributorAuthor

Hi @cmwilson21! Created a PR for this issue. Please check it out when chanced!

@github-actions

github-actionsBot commented Feb 28, 2023

Copy link
Copy Markdown
Contributor

Automatically generated comment ℹ️

This comment is automatically generated and will be overwritten every time changes are committed to this branch.

The table contains an overview of files in the content directory that have been changed in this pull request. It's provided to make it easy to review your changes on the staging site. Please note that changes to the data directory will not show up in this table.


Content directory changes

You may find it useful to copy this table into the pull request summary. There you can edit it to share links to important articles or changes and to give a high-level overview of how the changes in your pull request support the overall goals of the pull request.

SourcePreviewProductionWhat Changed
repositories/managing-your-repositorys-settings-and-features/customizing-your-repository/about-code-owners.mdfpt
ghec
ghes@ 3.83.73.63.53.4
ghae
fpt
ghec
ghes@ 3.83.73.63.53.4
ghae

fpt: Free, Pro, Team
ghec: GitHub Enterprise Cloud
ghes: GitHub Enterprise Server
ghae: GitHub AE

@cmwilson21

Copy link
Copy Markdown
Contributor

@OlayinkaAtobiloye awesome! I'll get this triaged for review! Thank you ✨

@cmwilson21cmwilson21 added content This issue or pull request belongs to the Docs Content team waiting for review Issue/PR is waiting for a writer's review code owners Content related to code owners repositories Content related to repositories and removed triage Do not begin working on this issue until triaged by the team labels Mar 1, 2023

@jc-clarkjc-clark left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This looks great to me @OlayinkaAtobiloye! Thank you for adding this ✨

@cmwilson21cmwilson21 added the ready to merge This pull request is ready to merge label Mar 2, 2023
@cmwilson21
cmwilson21 enabled auto-merge March 2, 2023 16:08
@cmwilson21
cmwilson21 merged commit 5a5e6cd into github:mainMar 2, 2023
@github-actions

Copy link
Copy Markdown
Contributor

Thanks very much for contributing! Your pull request has been merged 🎉 You should see your changes appear on the site in approximately 24 hours. If you're looking for your next contribution, check out our help wanted issues

Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

code ownersContent related to code ownerscontentThis issue or pull request belongs to the Docs Content teamready to mergeThis pull request is ready to mergerepositoriesContent related to repositorieswaiting for reviewIssue/PR is waiting for a writer's review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

CODEOWNERS file syntax documentation does not make it explicitly clear that a pattern can only be used once

4 participants

@OlayinkaAtobiloye@cmwilson21@jc-clark@docubot