Skip to content

✨ Bump k8s.io packages to v1.24.2 - #1940

Merged
k8s-ci-robot merged 2 commits into
kubernetes-sigs:release-0.12from
k8s-infra-cherrypick-robot:cherry-pick-1935-to-release-0.12
Jun 23, 2022
Merged

k8s-ci-robot merged 2 commits into
kubernetes-sigs:release-0.12from
k8s-infra-cherrypick-robot:cherry-pick-1935-to-release-0.12

Conversation

@k8s-infra-cherrypick-robot

Copy link
Copy Markdown

This is an automated cherry-pick of #1935

/assign FillZpp

@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Jun 23, 2022
@k8s-ci-robot
k8s-ci-robot requested review from alenkacz and gerred June 23, 2022 13:44
@k8s-ci-robot k8s-ci-robot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Jun 23, 2022
@sbueringer

sbueringer commented Jun 23, 2022

Copy link
Copy Markdown
Member

PR description is too short. Just worked around it in CAPI by not running this check against cherry-pick PRs: https://github.com/kubernetes-sigs/cluster-api/pull/6705/files

(new version of the kubebuilder release tools changed the check)

@sbueringer

Copy link
Copy Markdown
Member

Otherwise
/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 23, 2022
@alvaroaleman alvaroaleman added the tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges. label Jun 23, 2022
@alvaroaleman alvaroaleman changed the title [release-0.12] ✨ Bump k8s.io packages to v1.24.2 ✨ Bump k8s.io packages to v1.24.2 Jun 23, 2022
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: alvaroaleman, k8s-infra-cherrypick-robot

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 23, 2022
@k8s-ci-robot
k8s-ci-robot merged commit f561596 into kubernetes-sigs:release-0.12 Jun 23, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants