Skip to content

🌱 Use runtime.Unstructured interface instead of Unstructured struct - #2168

Merged
k8s-ci-robot merged 1 commit into
kubernetes-sigs:masterfrom
vincepri:use-runtime-unstructured
Feb 3, 2023
Merged

k8s-ci-robot merged 1 commit into
kubernetes-sigs:masterfrom
vincepri:use-runtime-unstructured

Conversation

@vincepri

@vincepri vincepri commented Feb 2, 2023

Copy link
Copy Markdown
Member

Signed-off-by: Vince Prignano vincepri@redhat.com

Fixes #415

I'm indifferent if we actually want to do this or not, if we decide not to, we should also close the related issue as well

Signed-off-by: Vince Prignano <vincepri@redhat.com>
@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Feb 2, 2023
@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 2, 2023
@vincepri vincepri changed the title Use runtime.Unstructured interface instead of Unstructured struct 🌱 Use runtime.Unstructured interface instead of Unstructured struct Feb 2, 2023
@k8s-ci-robot k8s-ci-robot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Feb 2, 2023
@vincepri

vincepri commented Feb 2, 2023

Copy link
Copy Markdown
Member Author

/assign @alvaroaleman @sbueringer

@FillZpp

FillZpp commented Feb 3, 2023

Copy link
Copy Markdown
Contributor

I'm indifferent if we actually want to do this or not, if we decide not to, we should also close the related issue as well

I'm in favor of this. The code looks more clear and we don't have to check both Unstructured and UnstructuredList in some of the places.

@alvaroaleman alvaroaleman left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks like a clear simplification to me and doesn't result in any public api changes or test changes

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 3, 2023
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: alvaroaleman, vincepri

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [alvaroaleman,vincepri]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

unstructured client should key off of the unstructured interface

5 participants