Skip to content

✨ Export DepthMetricWithPriority - #3597

Open
alvaroaleman wants to merge 1 commit into
kubernetes-sigs:mainfrom
alvaroaleman:export
Open

alvaroaleman wants to merge 1 commit into
kubernetes-sigs:mainfrom
alvaroaleman:export

Conversation

@alvaroaleman

Copy link
Copy Markdown
Member

Currently, it is impossible to get the priority dimension with a custom metric provider, because the internal MetricsProviderWithPriority references the internal DepthMetricWithPriority, making it impossible to implement the whole thing outside of controller-runtime itself.

/assign sburinger
/hold

Currently, it is impossible to get the priority dimension with a custom
metric provider, because the internal MetricsProviderWithPriority references the
internal DepthMetricWithPriority, making it impossible to implement the
whole thing outside of controller-runtime itself.
@kubernetes-prow kubernetes-prow Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 14, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

@alvaroaleman: GitHub didn't allow me to assign the following users: sburinger.

Note that only kubernetes-sigs members with read permissions, repo collaborators and people who have commented on this issue/PR can be assigned. Additionally, issues/PRs can only have 10 assignees at the same time.
For more information please see the contributor guide

Details

In response to this:

Currently, it is impossible to get the priority dimension with a custom metric provider, because the internal MetricsProviderWithPriority references the internal DepthMetricWithPriority, making it impossible to implement the whole thing outside of controller-runtime itself.

/assign sburinger
/hold

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: alvaroaleman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 14, 2026
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Sep 14, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants