Skip to content

add deprecate note on gen-resourcesdocs README and makefile - #445

Merged
k8s-ci-robot merged 1 commit into
kubernetes-sigs:masterfrom
Caesarsage:deprecate_gen-resourcesdocs
May 5, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes-sigs:masterfrom
Caesarsage:deprecate_gen-resourcesdocs

Conversation

@Caesarsage

Copy link
Copy Markdown
Contributor

Add deprecation note on gen-resourcesdocs.

Follows
deprecate -> announce -> wait -> remove (if needed)

Closes: #434

@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label May 4, 2026
@k8s-ci-robot k8s-ci-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label May 4, 2026
@tengqm

tengqm commented May 5, 2026

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label May 5, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Caesarsage, tengqm

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 5, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 4c4789d into kubernetes-sigs:master May 5, 2026
2 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Merge gen-resourcesdocs into gen-apidocs

3 participants