Skip to content

Update AudioSwitch to handle potential exception when unregistering audio device listeners - #944

Merged
davidliu merged 1 commit into
mainfrom
dl/audioswitch_commdevice_exception
May 18, 2026
Merged

davidliu merged 1 commit into
mainfrom
dl/audioswitch_commdevice_exception

Conversation

@davidliu

Copy link
Copy Markdown
Contributor

No description provided.

@changeset-bot

changeset-bot Bot commented May 18, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 40436f7

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
client-sdk-android Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

Dependency diff:

-\--- com.github.davidliu:audioswitch:70efa204dda4f468f989b4b3e1b5ecb6ffe56ceb
+\--- com.github.davidliu:audioswitch:039a35aefab7747c557242fa216c9ea11743b604

@github-actions

Copy link
Copy Markdown
Contributor

Diffuse output:

OLD: diffuse-source-file
NEW: livekit-android-sdk-release.aar

 AAR      │ old      │ new      │ diff 
──────────┼──────────┼──────────┼──────
      jar │    4 MiB │    4 MiB │  0 B 
 manifest │  1.5 KiB │  1.5 KiB │  0 B 
 lint-jar │ 12.7 KiB │ 12.7 KiB │  0 B 
    other │  1.8 KiB │  1.8 KiB │  0 B 
──────────┼──────────┼──────────┼──────
    total │    4 MiB │    4 MiB │  0 B 

 JAR     │ old   │ new   │ diff      
─────────┼───────┼───────┼───────────
 classes │  2215 │  2215 │ 0 (+0 -0) 
 methods │ 42995 │ 42995 │ 0 (+0 -0) 
  fields │  8403 │  8403 │ 0 (+0 -0)
AAR
 size │ diff │ path    
──────┼──────┼─────────
  0 B │  0 B │ (total)

@davidliu
davidliu merged commit c8ebe98 into main May 18, 2026
7 checks passed
@davidliu
davidliu deleted the dl/audioswitch_commdevice_exception branch May 18, 2026 17:03
@davidliu davidliu mentioned this pull request May 18, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant