Skip to content

add flexfec support - #1805

Merged
chenosaurus merged 2 commits into
mainfrom
dc/add_flexfec
Sep 18, 2026
Merged

chenosaurus merged 2 commits into
mainfrom
dc/add_flexfec

Conversation

@chenosaurus

Copy link
Copy Markdown
Contributor

No description provided.

@changeset-bot

changeset-bot Bot commented Sep 17, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 30e286f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 2 packages
Name Type
github.com/livekit/protocol Patch
@livekit/protocol Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@chenosaurus
chenosaurus merged commit 5290541 into main Sep 18, 2026
9 checks passed
@chenosaurus
chenosaurus deleted the dc/add_flexfec branch September 18, 2026 03:26
@github-actions github-actions Bot mentioned this pull request Sep 18, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants