Skip to content

fix: correct hydration for instant search - #1866

Merged
danielroe merged 1 commit into
npmx-dev:knowler/instant-search-advisoryfrom
alexdln:fix/instant-search-hydration
Mar 3, 2026
Merged

fix: correct hydration for instant search#1866
danielroe merged 1 commit into
npmx-dev:knowler/instant-search-advisoryfrom
alexdln:fix/instant-search-hydration

Conversation

@alexdln

@alexdlnalexdln commented Mar 3, 2026

Copy link
Copy Markdown
Member

PR to another branch (from #1845)

🧭 Context

image

@vercel

vercelBot commented Mar 3, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
npmx.devReadyReadyPreview, CommentMar 3, 2026 8:01am
2 Skipped Deployments
ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devIgnoredIgnoredPreviewMar 3, 2026 8:01am
npmx-lunariaIgnoredIgnoredMar 3, 2026 8:01am

Request Review

@danielroe
danielroe merged commit 824e152 into npmx-dev:knowler/instant-search-advisoryMar 3, 2026
12 checks passed
@alexdln
alexdln deleted the fix/instant-search-hydration branch March 3, 2026 09:40
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@alexdln@danielroe