Skip to content

chore(deps): update devdependency eslint-plugin-regexp to v3.1.0 - #2004

Merged
danielroe merged 1 commit into
mainfrom
renovate/all-linters
Mar 9, 2026
Merged

chore(deps): update devdependency eslint-plugin-regexp to v3.1.0#2004
danielroe merged 1 commit into
mainfrom
renovate/all-linters

Conversation

@renovate

@renovaterenovateBot commented Mar 9, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

PackageChangeAgeConfidence
eslint-plugin-regexp3.0.03.1.0ageconfidence

Warning

Some dependencies could not be looked up. Check the Dependency Dashboard for more information.


Release Notes

ota-meshi/eslint-plugin-regexp (eslint-plugin-regexp)

v3.1.0

Compare Source

Minor Changes
  • refactor: Name the default export 'regexp' (#​952)

Configuration

📅 Schedule: Branch creation - "on Monday" (UTC), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@vercel

vercelBot commented Mar 9, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devReadyReadyPreview, CommentMar 9, 2026 6:38am
npmx.devReadyReadyPreview, CommentMar 9, 2026 6:38am
1 Skipped Deployment
ProjectDeploymentActionsUpdated (UTC)
npmx-lunariaIgnoredIgnoredMar 9, 2026 6:38am

Request Review

@codecov

codecovBot commented Mar 9, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ All tests successful. No failed tests found.

📢 Thoughts on this report? Let us know!

@danielroe
danielroe merged commit 2f39f4e into mainMar 9, 2026
24 checks passed
@danielroe
danielroe deleted the renovate/all-linters branch March 9, 2026 07:59
@github-actionsgithub-actionsBot mentioned this pull request Mar 9, 2026
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@danielroe