Skip to content

chore(deps): upgrade to Vite 8 stable, Vite+ alpha, Vitest 4.1 - #2076

Closed
serhalp wants to merge 1 commit into
mainfrom
serhalp/vite8-upgrade
Closed

chore(deps): upgrade to Vite 8 stable, Vite+ alpha, Vitest 4.1#2076
serhalp wants to merge 1 commit into
mainfrom
serhalp/vite8-upgrade

Conversation

@serhalp

Copy link
Copy Markdown
Member

🔗 Linked issue

N/A

🧭 Context

These were just released on Thursday. See https://vite.dev/blog/announcing-vite8 and https://voidzero.dev/posts/announcing-vite-plus-alpha.

📚 Description

  • Vite 8 stable
  • Vite+ alpha
  • Vitest 4.1

@vercel

vercelBot commented Mar 14, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devReadyReadyPreview, CommentMar 14, 2026 3:40pm
npmx.devReadyReadyPreview, CommentMar 14, 2026 3:40pm
1 Skipped Deployment
ProjectDeploymentActionsUpdated (UTC)
npmx-lunariaIgnoredIgnoredMar 14, 2026 3:40pm

Request Review

@ghostdevv

Copy link
Copy Markdown
Member

I think this was done in #2129 so will close, let me know if I'm wrong 🙏

Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@serhalp@ghostdevv