Skip to content

chore(deps): update dependency std-env to v4 - #2095

Merged
danielroe merged 1 commit into
mainfrom
renovate/std-env-4.x
Mar 16, 2026
Merged

chore(deps): update dependency std-env to v4#2095
danielroe merged 1 commit into
mainfrom
renovate/std-env-4.x

Conversation

@renovate

@renovaterenovateBot commented Mar 16, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

PackageChangeAgeConfidence
std-env3.10.04.0.0ageconfidence

Warning

Some dependencies could not be looked up. Check the Dependency Dashboard for more information.


Release Notes

unjs/std-env (std-env)

v4.0.0

Compare Source

compare changes

🚀 Enhancements
  • Agent detection (#​179)
  • Check MODE variable for isDevelopment and isProduction flags (#​114)
🔥 Performance
🩹 Fixes
💅 Refactors
  • ⚠️ Remove env and process shims (d4de8a4)
  • ⚠️ Improve codebase (653b127)
  • Default nodeEnv to undefined (c78098a)
  • Remove toBoolean util (23bfb8f)
📦 Build
🏡 Chore
⚠️ Breaking Changes
  • ⚠️ Remove env and process shims (d4de8a4)
  • ⚠️ Improve codebase (653b127)
  • ⚠️ Esm-only dist (d31cca3)
❤️ Contributors

Configuration

📅 Schedule: Branch creation - "on Monday" (UTC), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@vercel

vercelBot commented Mar 16, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devReadyReadyPreview, CommentMar 16, 2026 6:11am
npmx.devReadyReadyPreview, CommentMar 16, 2026 6:11am
1 Skipped Deployment
ProjectDeploymentActionsUpdated (UTC)
npmx-lunariaIgnoredIgnoredMar 16, 2026 6:11am

Request Review

@codecov

codecovBot commented Mar 16, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ All tests successful. No failed tests found.

📢 Thoughts on this report? Let us know!

@danielroe
danielroe merged commit e145c14 into mainMar 16, 2026
24 checks passed
@danielroe
danielroe deleted the renovate/std-env-4.x branch March 16, 2026 10:02
@github-actionsgithub-actionsBot mentioned this pull request Mar 16, 2026
eryue0220 added a commit to eryue0220/npmx.dev that referenced this pull request Mar 17, 2026
…ix/issue-2044
* 'fix/issue-2044' of github.com:eryue0220/npmx.dev:
chore(deps): update dependency std-env to v4 (npmx-dev#2095)
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@danielroe