Skip to content

chore(deps): update devdependency @e18e/eslint-plugin to v0.3.0 - #2222

Merged
ghostdevv merged 1 commit into
mainfrom
renovate/all-linters
Mar 23, 2026
Merged

chore(deps): update devdependency @e18e/eslint-plugin to v0.3.0#2222
ghostdevv merged 1 commit into
mainfrom
renovate/all-linters

Conversation

@renovate

@renovaterenovateBot commented Mar 23, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

PackageChangeAgeConfidence
@e18e/eslint-plugin0.2.00.3.0ageconfidence

Warning

Some dependencies could not be looked up. Check the Dependency Dashboard for more information.


Release Notes

e18e/eslint-plugin (@​e18e/eslint-plugin)

v0.3.0

Compare Source

What's Changed

New Contributors

Full Changelog: e18e/eslint-plugin@0.2.0...0.3.0


Configuration

📅 Schedule: Branch creation - "on Monday" (UTC), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@vercel

vercelBot commented Mar 23, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devReadyReadyPreview, CommentMar 23, 2026 2:57am
npmx.devReadyReadyPreview, CommentMar 23, 2026 2:57am
1 Skipped Deployment
ProjectDeploymentActionsUpdated (UTC)
npmx-lunariaIgnoredIgnoredMar 23, 2026 2:57am

Request Review

@codecov

codecovBot commented Mar 23, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ All tests successful. No failed tests found.

📢 Thoughts on this report? Let us know!

@ghostdevv
ghostdevv added this pull request to the merge queueMar 23, 2026
Merged via the queue into main with commit 7083d6aMar 23, 2026
22 of 23 checks passed
@ghostdevv
ghostdevv deleted the renovate/all-linters branch March 23, 2026 03:03
@github-actionsgithub-actionsBot mentioned this pull request Mar 23, 2026
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@ghostdevv