Skip to content

chore(deps): lock file maintenance - #2224

Merged
ghostdevv merged 1 commit into
mainfrom
renovate/lock-file-maintenance
Mar 23, 2026
Merged

chore(deps): lock file maintenance#2224
ghostdevv merged 1 commit into
mainfrom
renovate/lock-file-maintenance

Conversation

@renovate

@renovaterenovateBot commented Mar 23, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

UpdateChange
lockFileMaintenanceAll locks refreshed

Warning

Some dependencies could not be looked up. Check the Dependency Dashboard for more information.

🔧 This Pull Request updates lock files to use the latest dependency versions.


Configuration

📅 Schedule: Branch creation - Between 12:00 AM and 03:59 AM, only on Monday ( * 0-3 * * 1 ) (UTC), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@vercel

vercelBot commented Mar 23, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devReadyReadyPreview, CommentMar 30, 2026 1:05am
npmx.devReadyReadyPreview, CommentMar 30, 2026 1:05am
1 Skipped Deployment
ProjectDeploymentActionsUpdated (UTC)
npmx-lunariaIgnoredIgnoredMar 30, 2026 1:05am

Request Review

@codecov

codecovBot commented Mar 23, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ All tests successful. No failed tests found.

📢 Thoughts on this report? Let us know!

@ghostdevv
ghostdevv added this pull request to the merge queueMar 23, 2026
Merged via the queue into main with commit c58ac2eMar 23, 2026
22 of 23 checks passed
@ghostdevv
ghostdevv deleted the renovate/lock-file-maintenance branch March 23, 2026 03:43
@github-actionsgithub-actionsBot mentioned this pull request Mar 23, 2026
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@ghostdevv