Skip to content

chore(deps): update devdependency typescript to v6 - #2242

Merged
danielroe merged 1 commit into
mainfrom
renovate/typescript-6.x
Mar 23, 2026
Merged

chore(deps): update devdependency typescript to v6#2242
danielroe merged 1 commit into
mainfrom
renovate/typescript-6.x

Conversation

@renovate

@renovaterenovateBot commented Mar 23, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

PackageChangeAgeConfidence
typescript (source)5.9.36.0.2ageconfidence

Warning

Some dependencies could not be looked up. Check the Dependency Dashboard for more information.


Release Notes

microsoft/TypeScript (typescript)

v6.0.2

Compare Source


Configuration

📅 Schedule: Branch creation - "on Monday" (UTC), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@vercel

vercelBot commented Mar 23, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devReadyReadyPreview, CommentMar 23, 2026 5:03pm
npmx.devReadyReadyPreview, CommentMar 23, 2026 5:03pm
1 Skipped Deployment
ProjectDeploymentActionsUpdated (UTC)
npmx-lunariaIgnoredIgnoredMar 23, 2026 5:03pm

Request Review

@codecov

codecovBot commented Mar 23, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ All tests successful. No failed tests found.

📢 Thoughts on this report? Let us know!

@danielroe
danielroe merged commit 223ffa4 into mainMar 23, 2026
23 checks passed
@danielroe
danielroe deleted the renovate/typescript-6.x branch March 23, 2026 16:57
@github-actionsgithub-actionsBot mentioned this pull request Mar 23, 2026
ghostdevv pushed a commit that referenced this pull request Mar 29, 2026
Co-authored-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@danielroe