Skip to content

chore(deps): update chromaui/action action to v16 - #2243

Merged
ghostdevv merged 1 commit into
mainfrom
renovate/chromaui-action-16.x
Mar 23, 2026
Merged

chore(deps): update chromaui/action action to v16#2243
ghostdevv merged 1 commit into
mainfrom
renovate/chromaui-action-16.x

Conversation

@renovate

@renovaterenovateBot commented Mar 23, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

PackageTypeUpdateChange
chromaui/actionactionmajorv15.3.0v16.0.0

Warning

Some dependencies could not be looked up. Check the Dependency Dashboard for more information.


Release Notes

chromaui/action (chromaui/action)

v16.0.0

Compare Source

💥 Breaking Change
Authors: 1

v15.3.1

Compare Source

🐛 Bug Fix
Authors: 1


Configuration

📅 Schedule: Branch creation - "on Monday" (UTC), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@vercel

vercelBot commented Mar 23, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
npmx.devReadyReadyPreview, CommentMar 23, 2026 5:08pm
2 Skipped Deployments
ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devIgnoredIgnoredPreviewMar 23, 2026 5:08pm
npmx-lunariaIgnoredIgnoredMar 23, 2026 5:08pm

Request Review

@codecov

codecovBot commented Mar 23, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ All tests successful. No failed tests found.

📢 Thoughts on this report? Let us know!

@ghostdevv
ghostdevv added this pull request to the merge queueMar 23, 2026
Merged via the queue into main with commit e3b575aMar 23, 2026
23 checks passed
@ghostdevv
ghostdevv deleted the renovate/chromaui-action-16.x branch March 23, 2026 17:15
@github-actionsgithub-actionsBot mentioned this pull request Mar 23, 2026
ghostdevv pushed a commit that referenced this pull request Mar 29, 2026
Co-authored-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@ghostdevv