Skip to content

chore(deps): update devdependency @vitest/coverage-v8 to v4.1.2 - #2322

Merged
danielroe merged 1 commit into
mainfrom
renovate/test-packages
Mar 31, 2026
Merged

chore(deps): update devdependency @vitest/coverage-v8 to v4.1.2#2322
danielroe merged 1 commit into
mainfrom
renovate/test-packages

Conversation

@renovate

@renovaterenovateBot commented Mar 30, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

PackageChangeAgeConfidence
@vitest/coverage-v8 (source)4.1.14.1.2ageconfidence

Warning

Some dependencies could not be looked up. Check the Dependency Dashboard for more information.


Release Notes

vitest-dev/vitest (@​vitest/coverage-v8)

v4.1.2

Compare Source

This release bumps Vitest's flatted version and removes version pinning to resolve flatted's CVE related issues (#​9975).

🐞 Bug Fixes
View changes on GitHub

Configuration

📅 Schedule: Branch creation - "on Monday" (UTC), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@vercel

vercelBot commented Mar 30, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devReadyReadyPreview, CommentMar 31, 2026 11:11am
npmx.devReadyReadyPreview, CommentMar 31, 2026 11:11am
1 Skipped Deployment
ProjectDeploymentActionsUpdated (UTC)
npmx-lunariaIgnoredIgnoredMar 31, 2026 11:11am

Request Review

@codecov

codecovBot commented Mar 30, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ All tests successful. No failed tests found.

📢 Thoughts on this report? Let us know!

@danielroe
danielroe merged commit 934181b into mainMar 31, 2026
23 checks passed
@danielroe
danielroe deleted the renovate/test-packages branch March 31, 2026 13:34
@github-actionsgithub-actionsBot mentioned this pull request Mar 31, 2026
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@danielroe