Skip to content

chore(deps): update dependency vite-plugin-pwa to v1.3.0 - #2706

Merged
danielroe merged 1 commit into
mainfrom
renovate/vite
May 11, 2026
Merged

chore(deps): update dependency vite-plugin-pwa to v1.3.0#2706
danielroe merged 1 commit into
mainfrom
renovate/vite

Conversation

@renovate

@renovaterenovateBot commented May 11, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

PackageChangeAgeConfidence
vite-plugin-pwa1.2.01.3.0ageconfidence

Release Notes

vite-pwa/vite-plugin-pwa (vite-plugin-pwa)

v1.3.0

Compare Source

🚀 Features
View changes on GitHub

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • "on Monday"
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@vercel

vercelBot commented May 11, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devReadyReadyPreview, CommentMay 11, 2026 0:59am
npmx.devReadyReadyPreview, CommentMay 11, 2026 0:59am
1 Skipped Deployment
ProjectDeploymentActionsUpdated (UTC)
npmx-lunariaIgnoredIgnoredMay 11, 2026 0:59am

Request Review

@github-actions

Copy link
Copy Markdown

e18e dependency analysis

No dependency warnings found.

@codecov

codecovBot commented May 11, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ All tests successful. No failed tests found.

📢 Thoughts on this report? Let us know!

@danielroe
danielroe merged commit b72ad4e into mainMay 11, 2026
28 checks passed
@danielroe
danielroe deleted the renovate/vite branch May 11, 2026 07:35
@github-actionsgithub-actionsBot mentioned this pull request May 11, 2026
ayo-run pushed a commit to ayo-run/npmx.dev that referenced this pull request Aug 5, 2026
Co-authored-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@danielroe@gameroman