Skip to content

feat: add nodejs noodle (#2778) - #2801

Closed
alexdln wants to merge 1 commit into
npmx-dev:releasefrom
alexdln:chore/nodejs-noodle
Closed

feat: add nodejs noodle (#2778)#2801
alexdln wants to merge 1 commit into
npmx-dev:releasefrom
alexdln:chore/nodejs-noodle

Conversation

@alexdln

Copy link
Copy Markdown
Member

🧭 Context

A new noodle that was added in #2778. Merging it into the release before the event

image

@alexdln
alexdln requested review from a teamMay 26, 2026 10:39
@vercel

vercelBot commented May 26, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
npmx.devReadyReadyPreview, CommentMay 26, 2026 2:00pm
2 Skipped Deployments
ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devIgnoredIgnoredPreviewMay 26, 2026 2:00pm
npmx-lunariaIgnoredIgnoredMay 26, 2026 2:00pm

Request Review

@coderabbitai

coderabbitaiBot commented May 26, 2026

Copy link
Copy Markdown
Contributor

Important

Review skipped

Auto reviews are disabled on base/target branches other than the default branch.

Please check the settings in the CodeRabbit UI or the .coderabbit.yaml file in this repository. To trigger a single review, invoke the @coderabbitai review command.

⚙️ Run configuration

Configuration used: Path: .coderabbit.yaml

Review profile: CHILL

Plan: Pro

Run ID: 087a7a8a-e54b-4f9e-9329-cd5d41e13535

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review
✨ Finishing Touches
🧪 Generate unit tests (beta)
  • Create PR with unit tests

Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out.

❤️ Share

Comment @coderabbitai help to get the list of available commands and usage tips.

@danielroe
danielroeforce-pushed the chore/nodejs-noodle branch from b644514 to 05a8324CompareMay 26, 2026 13:57
@danielroe

Copy link
Copy Markdown
Member

pushed with no changes to hopefully trigger ci

@patak-catpatak-cat left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

💚

@patak-cat

Copy link
Copy Markdown
Contributor

Closing this one, as main was directly merged as 0.13

Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@alexdln@danielroe@patak-cat@gameroman