Uh oh!
There was an error while loading. Please reload this page.
chore(deps): update dependency @nuxt/test-utils to v4.2.0 - #3215
Open
renovate[bot] wants to merge 1 commit into
Open
chore(deps): update dependency @nuxt/test-utils to v4.2.0#3215renovate[bot] wants to merge 1 commit into
renovate[bot] wants to merge 1 commit into
Conversation
Contributor
The latest updates on your projects. Learn more about Vercel for GitHub.
2 Skipped Deployments
|
ContributorAuthor
|
e18e dependency analysisNo dependency warnings found. |
|
renovateBotforce-pushed
the
renovate/nuxt
branch
from
August 28, 2026 19:39
f8fbf48 to
c6bee0dComparerenovateBotforce-pushed
the
renovate/nuxt
branch
from
August 29, 2026 10:22
c6bee0d to
d728381ComparerenovateBotforce-pushed
the
renovate/nuxt
branch
from
August 29, 2026 10:29
d728381 to
63cbe37ComparerenovateBotforce-pushed
the
renovate/nuxt
branch
from
August 31, 2026 01:00
63cbe37 to
0c20411ComparerenovateBotforce-pushed
the
renovate/nuxt
branch
from
August 31, 2026 06:56
0c20411 to
8a2c827ComparerenovateBotforce-pushed
the
renovate/nuxt
branch
from
August 31, 2026 09:15
8a2c827 to
defa07bComparerenovateBotforce-pushed
the
renovate/nuxt
branch
from
September 5, 2026 14:33
defa07b to
e6a8636ComparerenovateBotforce-pushed
the
renovate/nuxt
branch
from
September 5, 2026 14:39
e6a8636 to
081a2edComparerenovateBotforce-pushed
the
renovate/nuxt
branch
from
September 6, 2026 12:43
081a2ed to
35a7bd8ComparerenovateBotforce-pushed
the
renovate/nuxt
branch
from
September 6, 2026 12:50
35a7bd8 to
4ab0833Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for freeto join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
4.1.0→4.2.0Release Notes
nuxt/test-utils (@nuxt/test-utils)
v4.2.0Compare Source
👉 Changelog
compare changes
🚀 Enhancements
🩹 Fixes
mockNuxtImportsourcemap mapped position (#1787)setupTimeoutin playwright + surface server logs (#1785)includeSourcein both projects indefineVitestConfig(#1763)taskkillto kill processes on windows (#1760)🏡 Chore
latestranges with caret ranges (#1773)✅ Tests
🤖 CI
🎉 New Contributors
❤️ Contributors
Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.