Skip to content

Remove popover responsive hack - #1759

Merged
langermank merged 7 commits into
mainfrom
hx-popover-remedy
Nov 22, 2021
Merged

Remove popover responsive hack#1759
langermank merged 7 commits into
mainfrom
hx-popover-remedy

Conversation

@langermank

Copy link
Copy Markdown
Contributor
  • Copies Popoverhack into Primer CSS, removing .page-responsive
  • By default, on screens smaller than 767pxPopover will be full-width and without caret
  • Removed dependency on Box
  • Organized docs a bit

Fixes: https://github.com/github/primer/issues/324

/cc @primer/css-reviewers

@langermank
langermank requested a review from a team as a code ownerNovember 19, 2021 19:17
@changeset-bot

changeset-botBot commented Nov 19, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ac124d2

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/cssMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@jonrohanjonrohan left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks good, and Awesome that you're adding stories as you're going ✨

Comment on lines +202 to +205
top: auto;
right: 0;
bottom: 0;
left: 0;

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@langermank Sorry this is too late and would need to be a follow-up PR.

But we might wanna keep the !important. There are cases where Popovers are positioned manually to adjust to certain content. E.g. using right: -327px; bottom: -8px.

Also the width of the Popover-message that probably gets customized too. E.g. width: 250px;

Copy link
Copy Markdown
ContributorAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Ah! No problem, will follow up with a fix. Thanks :)

Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@langermank@jonrohan@simurai