Skip to content

Use counter-border for LHC - #1792

Merged
simurai merged 4 commits into
mainfrom
counter-border
Dec 7, 2021
Merged

Use counter-border for LHC#1792
simurai merged 4 commits into
mainfrom
counter-border

Conversation

@simurai

@simuraisimurai commented Dec 1, 2021

Copy link
Copy Markdown
Contributor

This is a follow-up to primer/primitives#279. It uses --color-counter-border for the .Counter border. It should improve contrast for the LHC (Light High Contrast) theme:

BeforeAfter
Screen Shot 2021-12-01 at 15 14 08Screen Shot 2021-12-01 at 15 14 51

Other themes will keep using transparent.

TODO

@simurai
simurai requested a review from a team as a code ownerDecember 1, 2021 06:16
@changeset-bot

changeset-botBot commented Dec 1, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 98d24dd

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/cssPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@simurai
simurai marked this pull request as draft December 1, 2021 06:20
@simurai
simurai merged commit b1c43f1 into mainDec 7, 2021
@simurai
simurai deleted the counter-border branch December 7, 2021 05:56
@primer-cssprimer-css mentioned this pull request Dec 7, 2021
jonrohan pushed a commit that referenced this pull request Dec 8, 2021
* Use counter-border for LHC
* Create orange-ties-sin.md
* Remove fallback
jonrohan added a commit that referenced this pull request Dec 8, 2021
* Updating and autofixing stylelint
* Moving config to primer/stylelint-config
* Removing unused disables from css
* Stylelint auto-fixes
* @primer/stylelint-config@12.2.0
* Remove these from workfow
* Use reusable release_canary workflow (#1811)
* Use reusable release_canary workflow
* Install with yarn
Co-authored-by: Jon Rohan <yes@jonrohan.codes>
* Use counter-border for LHC (#1792)
* Use counter-border for LHC
* Create orange-ties-sin.md
* Remove fallback
* UnderlineNav `:focus` styles (#1764)
* add pseudo selectors
* adjustments
* add stories, cleanup
* update mixin
* fix mixin
* lint
* add back overflow styles
* adjust focus for better overflow state, scrollsnap
* post test adjustments, move hacks to primer css
* Stylelint auto-fixes
* hover state desktop only
* document data-content hack
* Create nice-days-jog.md
Co-authored-by: Actions Auto Build <actions@github.com>
Co-authored-by: simurai <simurai@github.com>
* Add note about loading the "latest" Primer CSS version (#1784)
* Add stashing
* Stylelint auto-fixes
Co-authored-by: Actions Auto Build <actions@github.com>
Co-authored-by: Cole Bemis <colebemis@github.com>
Co-authored-by: simurai <simurai@github.com>
Co-authored-by: Katie Langerman <langermank@github.com>
jonrohan added a commit that referenced this pull request Dec 8, 2021
* add pseudo selectors
* adjustments
* add stories, cleanup
* update mixin
* fix mixin
* lint
* add back overflow styles
* adjust focus for better overflow state, scrollsnap
* post test adjustments, move hacks to primer css
* Stylelint auto-fixes
* hover state desktop only
* document data-content hack
* Create nice-days-jog.md
* Use counter-border for LHC (#1792)
* Use counter-border for LHC
* Create orange-ties-sin.md
* Remove fallback
* Update nice-days-jog.md
Co-authored-by: langermank <langermank@github.com>
Co-authored-by: Actions Auto Build <actions@github.com>
Co-authored-by: simurai <simurai@github.com>
jonrohan added a commit that referenced this pull request Apr 29, 2022
* Adding pre.json
* Delete pre.json
* Marketing: Remove unused pullquote selector (#1767)
* Remove unused pullquote style
* Create good-fireants-collect.md
* next_major protection
* On push
* Exit 1
* Update instructions
* Underline nav focus state (#1821)
* add pseudo selectors
* adjustments
* add stories, cleanup
* update mixin
* fix mixin
* lint
* add back overflow styles
* adjust focus for better overflow state, scrollsnap
* post test adjustments, move hacks to primer css
* Stylelint auto-fixes
* hover state desktop only
* document data-content hack
* Create nice-days-jog.md
* Use counter-border for LHC (#1792)
* Use counter-border for LHC
* Create orange-ties-sin.md
* Remove fallback
* Update nice-days-jog.md
Co-authored-by: langermank <langermank@github.com>
Co-authored-by: Actions Auto Build <actions@github.com>
Co-authored-by: simurai <simurai@github.com>
* Global focus styles (#1744)
* setup base focus styles
* add feature stories
* add stories for focusable things, delete outline:0
* switch back to box-shadow
* support class
* stylelint
* fix theme viewer
* switch back to outline, address feedback
* lint
* temp stashing stories here
* Create giant-bees-talk.md
* I think we got it!
* address input directly
* checkbox/radio outline offset
* lint
* change actionlist to just use focus
* merge
* Update giant-bees-talk.md
* address marketing styles
* tabnav focus fix
* reach all buttons
* attempt windows hc selector
* Stylelint auto-fixes
* fixes
* add focus style testing page
* Stylelint auto-fixes
* add href for testing
* remove position relative to fix chrome bug
* fix details scenario
* add offset to WHC
* maintain offset specificity in whc
* inset tabnav focus
* switch offset to inset
* fix actionlist focus
* lint
* better scoping, handle forms for safari
* moving specific styles from dotcom
* address autocomplete
* cleanup
* cleanup
* selected focus states
* adjust marketing focus
* use offset instead for marketing
* Stylelint auto-fixes
* fix merge
Co-authored-by: Jon Rohan <yes@jonrohan.codes>
Co-authored-by: Actions Auto Build <actions@github.com>
* Upgrading to stylelint-config-12.4.0 (#2041)
* Upgrading to stylelint-config-12.4.0
* Create tame-clouds-shop.md
* Global focus style CSS from feature flag (#2046)
* extra CSS found from cleaning up feat flag
* lint
* Create young-trainers-yell.md
* fix mixin
Co-authored-by: Jon Rohan <yes@jonrohan.codes>
* UnderlineNav bug fix (#2047)
* fix selector (bad merge?)
* Create ninety-bobcats-do.md
* lint
* Comment box upload focus border-radius (#2049)
* missing border-radius
* Create beige-horses-fetch.md
* lint
* you would think I would know these by now
* Fix headings with an anchor in a summary (#2048)
* Fix headings with an anchor in a summary
* Create hip-readers-press.md
* Lint
* Version Packages
Co-authored-by: Jon Rohan <yes@jonrohan.codes>
Co-authored-by: Tobias Ahlin <hello@tobiasahlin.com>
Co-authored-by: langermank <langermank@github.com>
Co-authored-by: Actions Auto Build <actions@github.com>
Co-authored-by: simurai <simurai@github.com>
Co-authored-by: github-actions[bot] <github-actions[bot]@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@simurai@Juliusschaeper