Skip to content

[Layout component] Avoid resizing main while loading - #1818

Merged
jonrohan merged 4 commits into
mainfrom
layout-width-loading-fix
Dec 9, 2021
Merged

[Layout component] Avoid resizing main while loading#1818
jonrohan merged 4 commits into
mainfrom
layout-width-loading-fix

Conversation

@vdepizzol

@vdepizzolvdepizzol commented Dec 7, 2021

Copy link
Copy Markdown
Contributor

This quick fix makes sure the alpha Layout component doesn't render the main column at full width while the page is loading.

/cc @jonrohan

@changeset-bot

changeset-botBot commented Dec 7, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 169b723

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/cssPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@jonrohan
jonrohan marked this pull request as ready for review December 8, 2021 23:32
@jonrohan
jonrohan requested a review from a team as a code ownerDecember 8, 2021 23:32
@jonrohan
jonrohan merged commit 22c29e7 into mainDec 9, 2021
@jonrohan
jonrohan deleted the layout-width-loading-fix branch December 9, 2021 00:04
@primer-cssprimer-css mentioned this pull request Dec 9, 2021
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@vdepizzol@jonrohan