Skip to content

Remove article - #2090

Merged
simurai merged 3 commits into
primer:mainfrom
Luckz:patch-1
Jul 26, 2022
Merged

Remove article#2090
simurai merged 3 commits into
primer:mainfrom
Luckz:patch-1

Conversation

@Luckz

Copy link
Copy Markdown
Contributor

The 'the' is superfluous / wrong. unpkg doesn't refer to itself as "the unpkg" or anything.

What are you trying to accomplish?

Minor grammar correction

Can these changes ship as is?

  • Yes, this PR does not depend on additional changes. 🚢

The 'the' is superfluous / wrong. unpkg doesn't refer to itself as "the unpkg" or anything.
@Luckz
Luckz requested a review from a team as a code ownerMay 21, 2022 09:06
@Luckz
Luckz requested a review from langermankMay 21, 2022 09:06
@changeset-bot

changeset-botBot commented May 21, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d178110

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/cssPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

Hi! This pull request has been marked as stale because it has been open with no activity for 60 days. You can comment on the pull request or remove the stale label to keep it open. If you do nothing, this pull request will be closed in 7 days.

@github-actionsgithub-actionsBot added the Stale Automatically marked as stale. label Jul 25, 2022
@Luckz

Copy link
Copy Markdown
ContributorAuthor

comment.

@github-actionsgithub-actionsBot removed the Stale Automatically marked as stale. label Jul 25, 2022

@simuraisimurai left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM! 🚢👍

@simurai
simurai merged commit 8016e3c into primer:mainJul 26, 2022
@primer-cssprimer-css mentioned this pull request Jul 26, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@Luckz@simurai