Skip to content

Prevent SegmentedControl-text from wrapping - #2174

Merged
langermank merged 3 commits into
mainfrom
segmented-control-wrapping
Jul 28, 2022
Merged

Prevent SegmentedControl-text from wrapping#2174
langermank merged 3 commits into
mainfrom
segmented-control-wrapping

Conversation

@simurai

@simuraisimurai commented Jul 27, 2022

Copy link
Copy Markdown
Contributor

What are you trying to accomplish?

Follow-up to #2083. It prevents text from wrapping onto multiple lines.

BeforeAfter
Screen Shot 2022-07-27 at 12 08 54Screen Shot 2022-07-27 at 12 17 04

What approach did you choose and why?

white-space: nowrap;

Can these changes ship as is?

  • Yes, this PR does not depend on additional changes. 🚢

@simurai
simurai requested a review from a team as a code ownerJuly 27, 2022 03:15
@simurai
simurai requested a review from langermankJuly 27, 2022 03:15
@changeset-bot

changeset-botBot commented Jul 27, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 115675b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/cssPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@simurai
simurai requested a review from jonrohanJuly 27, 2022 03:17
@github-actions
github-actionsBot temporarily deployed to Storybook Preview July 27, 2022 03:21 Inactive
@simurai
simurai temporarily deployed to github-pages July 27, 2022 03:23 Inactive
@github-actions
github-actionsBot temporarily deployed to Storybook Preview July 27, 2022 03:23 Inactive
@github-actions
github-actionsBot temporarily deployed to Storybook Preview July 28, 2022 01:13 Inactive
@langermank
langermank merged commit 97f178d into mainJul 28, 2022
@langermank
langermank deleted the segmented-control-wrapping branch July 28, 2022 01:14
@primer-cssprimer-css mentioned this pull request Jul 28, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@simurai@langermank