Skip to content

Upgrade primer/primitives to 4.4.0 - #1297

Merged
colebemis merged 7 commits into
mainfrom
VanAnderson/upgrade-primitives-to-latest
Jun 14, 2021
Merged

Upgrade primer/primitives to 4.4.0#1297
colebemis merged 7 commits into
mainfrom
VanAnderson/upgrade-primitives-to-latest

Conversation

@VanAnderson

Copy link
Copy Markdown
Contributor

This upgrades primer/primitives and can make some crucial design fixes.

@changeset-bot

changeset-botBot commented Jun 11, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b451448

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/componentsPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@vercel

vercelBot commented Jun 11, 2021

Copy link
Copy Markdown

This pull request is being automatically deployed with Vercel (learn more).
To see the status of your deployment, click below or on the icon next to each commit.

🔍 Inspect: https://vercel.com/primer/primer-components/A4CFZbqZhoUHqQS1xMKZU7jcuXCC
✅ Preview: https://primer-components-git-vananderson-upgrade-primiti-bba493-primer.vercel.app

Comment threadsrc/__tests__/__snapshots__/StateLabel.tsx.snap Outdated
@VanAnderson

Copy link
Copy Markdown
ContributorAuthor

image

Also a bit concerning

@colebemiscolebemis changed the title Upgrade primer/primitives to 4.3.11Upgrade primer/primitives to 4.4.0Jun 14, 2021
@colebemis

Copy link
Copy Markdown
Contributor

image

Also a bit concerning

Looks like this is happening on main too. I believe it's a result of using portals to render the action menu. Not a concern for this PR 👍

@colebemis
colebemis merged commit 5fb3683 into mainJun 14, 2021
@colebemis
colebemis deleted the VanAnderson/upgrade-primitives-to-latest branch June 14, 2021 18:44
@github-actionsgithub-actionsBot mentioned this pull request Jun 14, 2021
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@VanAnderson@colebemis@smockle@dusave