Skip to content

Add flex to timeline item - #1403

Merged
colebemis merged 3 commits into
primer:mainfrom
mathiasbosman:timeline.item_fix_flex
Sep 3, 2021
Merged

Add flex to timeline item#1403
colebemis merged 3 commits into
primer:mainfrom
mathiasbosman:timeline.item_fix_flex

Conversation

@mathiasbosman

Copy link
Copy Markdown
Contributor

Added display:flex; to the Timeline.Item component.

Closes#1402

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

@mathiasbosman
mathiasbosman requested a review from a teamSeptember 3, 2021 10:36
@changeset-bot

changeset-botBot commented Sep 3, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1123ed2

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/componentsPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mathiasbosmanmathiasbosman mentioned this pull request Sep 3, 2021
@colebemis

Copy link
Copy Markdown
Contributor

👋 Hi, @mathiasbosman! Thanks for the PR. Could you explain what problems display: flex fixes here?

@mathiasbosman

Copy link
Copy Markdown
ContributorAuthor

👋 Hi, @mathiasbosman! Thanks for the PR. Could you explain what problems display: flex fixes here?

I added a screenshot in the issue #1402. The content of the item is sticking to the timeline.

It's visible in the docs as well: https://primer.style/react/Timeline

Unless this is intended?

@colebemis

Copy link
Copy Markdown
Contributor

Ah, perfect. Thank you for fixing this!

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

❤️

@colebemis
colebemis enabled auto-merge (squash) September 3, 2021 18:38
@colebemis
colebemis merged commit 65cab2c into primer:mainSep 3, 2021
@primer-cssprimer-css mentioned this pull request Sep 3, 2021
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Timeline.Item alignment

2 participants

@mathiasbosman@colebemis