Skip to content

Fix auxiliaryText in ActionList/Header.tsx - #1549

Merged
siddharthkp merged 2 commits into
primer:mainfrom
michael-lefebvre:patch-1
Nov 1, 2021
Merged

Fix auxiliaryText in ActionList/Header.tsx#1549
siddharthkp merged 2 commits into
primer:mainfrom
michael-lefebvre:patch-1

Conversation

@michael-lefebvre

Copy link
Copy Markdown
Contributor

the auxiliaryText property are not displayed due to missing curly braces

Describe your changes here.

Closes # (type the issue number after # if applicable; otherwise remove this line)

Screenshots

Please provide before/after screenshots for any visual changes

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

the auxiliaryText property are not displayed due to missing curly braces
@michael-lefebvre
michael-lefebvre requested review from a team and rezrahNovember 1, 2021 16:17
@changeset-bot

changeset-botBot commented Nov 1, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 5a3d5f4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/componentsPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks for the fix! You can tell auxiliaryText isn't very popular 😄

@siddharthkp
siddharthkp merged commit f565840 into primer:mainNov 1, 2021
@primer-cssprimer-css mentioned this pull request Nov 1, 2021
@primer-cssprimer-css mentioned this pull request Nov 16, 2021
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@michael-lefebvre@siddharthkp