Skip to content

ActionList: Correct semantics for Groups - #1584

Closed
siddharthkp wants to merge 118 commits into
mainfrom
siddharth/action-list-groups-a11y
Closed

ActionList: Correct semantics for Groups#1584
siddharthkp wants to merge 118 commits into
mainfrom
siddharth/action-list-groups-a11y

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Nov 8, 2021

Copy link
Copy Markdown
Member

Part of https://github.com/github/primer/issues/447

  • ActionList.Group create a ul inside a li
  • The title of the Group is used as the label for the ul.
  • The section Header is for visual users only. It's hidden for screen readers.
  • Group can accept role

This is the generated HTML, formatted for legibility:

image

<ularia-label="Select reviewers"><li><divrole="presentation" aria-hidden="true"><spanid="react-aria-1">Suggestions</span></div><ularia-labelledby="react-aria-1" role="listbox"><lirole="option"
aria-selected="true"
aria-labelledby="react-aria-2"
aria-describedby="react-aria-3 react-aria-4"
><span><inputtype="checkbox" tabindex="-1" readonlyaria-readonly="false" checked/></span><spanclass="Box-nv15kw-0 iBXgEg"><imgsrc="https://github.com/pksjce.png" /></span><divdata-component="ActionList.Item--DividerContainer"><div><spanid="react-aria-2">pksjce</span><divid="react-aria-3" title="Pavithra Kodmad">Pavithra Kodmad</div></div><spanid="react-aria-4">Recently edited these files</span></div></li><li>....</li></ul></li><li><divrole="presentation" aria-hidden="true"><spanid="react-aria-5">Everyone</span></div><ularia-labelledby="react-aria-5" role="listbox"><li>...</li><li>...</li><li>...</li></ul></li>

Merge checklist

@changeset-bot

changeset-botBot commented Nov 8, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 75bec0a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/componentsPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkp
siddharthkp changed the base branch from main to action-list-2November 8, 2021 14:42
@github-actions

github-actionsBot commented Nov 8, 2021

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js54.44 KB (0%)
dist/browser.umd.js54.83 KB (0%)

@siddharthkpsiddharthkp changed the title [ActionList] Correct semantics for GroupsActionList: Correct semantics for GroupsNov 8, 2021
@siddharthkpsiddharthkp self-assigned this Nov 8, 2021
Base automatically changed from action-list-2 to mainNovember 11, 2021 13:32
@siddharthkp

Copy link
Copy Markdown
MemberAuthor

Moved changes over to #1598

@joshblack
joshblack deleted the siddharth/action-list-groups-a11y branch January 19, 2023 16:44
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@jfuchs