Skip to content

ActionList: Use icon instead of input for multiple selection - #1591

Closed
siddharthkp wants to merge 119 commits into
mainfrom
siddharthkp/actionlist-selection-a11y
Closed

ActionList: Use icon instead of input for multiple selection#1591
siddharthkp wants to merge 119 commits into
mainfrom
siddharthkp/actionlist-selection-a11y

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Nov 10, 2021

Copy link
Copy Markdown
Member

Change base after #1517 is merged
Part of https://github.com/github/primer/issues/447
Fixes half of https://github.com/github/primer/issues/228

Nested interactive controls are not announced by screen readers, the accessibility advice here is not use interactive elements for presentation.

  • Replace readonly input type="checkbox" with svg from primer/css
  • Bonus: checkmark consistency across browsers :)

Screenshots

look pretty:
image

Accessibility violations:

Before:

image

After:

image

Please provide before/after screenshots for any visual changes

Merge checklist

  • NA Added/updated tests
  • NA Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@changeset-bot

changeset-botBot commented Nov 10, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: fb551cb

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/componentsPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Nov 10, 2021

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js54.7 KB (+0.53% 🔺)
dist/browser.umd.js55.08 KB (+0.5% 🔺)

Base automatically changed from action-list-2 to mainNovember 11, 2021 13:32
@siddharthkp

Copy link
Copy Markdown
MemberAuthor

Replaced by #1601

@joshblack
joshblack deleted the siddharthkp/actionlist-selection-a11y branch January 19, 2023 16:44
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@jfuchs