Skip to content

Rename unreleased to drafts to avoid confusion - #1609

Merged
siddharthkp merged 3 commits into
mainfrom
siddharth/rename-unreleased-to-drafts
Nov 16, 2021
Merged

Rename unreleased to drafts to avoid confusion#1609
siddharthkp merged 3 commits into
mainfrom
siddharth/rename-unreleased-to-drafts

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Nov 15, 2021

Copy link
Copy Markdown
Member

Rename @primer/components/unreleased to @primer/components/drafts

--

Picked a name that doesn't conflict with component maturity status - experimental/alpha/beta/stable
& doesn't conflict with package publishing language - published/released/next/apha/beta

Thanks to @pksjce for the suggestion!

--

  • In the docs, because we have multiple ActionLists now, we import it ActionList2 from a different scope:
const{ActionList}=draftsrender(<ActionList>
...
</ActionList>)

This causes confusion in where does drafts come from?

The long term fix would be to modify the scope conditionally by tagging the code block with "drafts": ```javascript live drafts.

For now, the bandaid is the add a comment to ignore the internal silliness and import from the path

// import {ActionList} from '@primer/components/drafts'const{ActionList}=drafts// ignore docs silliness, import like that ↑

@siddharthkp
siddharthkp requested review from a team and rezrahNovember 15, 2021 17:05
@changeset-bot

changeset-botBot commented Nov 15, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 5e5f550

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/componentsPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkpsiddharthkp changed the title rename unreleased to draftsRename unreleased to drafts to avoid confusionNov 15, 2021
@siddharthkpsiddharthkp self-assigned this Nov 15, 2021
@siddharthkpsiddharthkp added docs Documentation react labels Nov 15, 2021
@siddharthkp

Copy link
Copy Markdown
MemberAuthor

@pksjce@jclem FYI!

@github-actions

github-actionsBot commented Nov 15, 2021

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js54.96 KB (0%)
dist/browser.umd.js55.35 KB (0%)

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

:shipit:

@siddharthkp
siddharthkp merged commit 5eb7ade into mainNov 16, 2021
@siddharthkp
siddharthkp deleted the siddharth/rename-unreleased-to-drafts branch November 16, 2021 12:04
This was referenced Nov 16, 2021
@siddharthkpsiddharthkp mentioned this pull request Dec 6, 2021
6 tasks
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

docsDocumentationreact

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@colebemis