Skip to content

ActionList v2: Visual update for disabled Items - #1629

Merged
siddharthkp merged 2 commits into
mainfrom
actionlist/disabled-actionlist-item
Nov 19, 2021
Merged

ActionList v2: Visual update for disabled Items#1629
siddharthkp merged 2 commits into
mainfrom
actionlist/disabled-actionlist-item

Conversation

@siddharthkp

Copy link
Copy Markdown
Member

Screenshots

Before:

image

After:

image

Merge checklist

  • NA Added/updated tests
  • NA Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

@siddharthkp
siddharthkp requested review from a team and colebemisNovember 19, 2021 15:40
@changeset-bot

changeset-botBot commented Nov 19, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: df271d4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/componentsPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Nov 19, 2021

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js56.31 KB (0%)
dist/browser.umd.js56.7 KB (0%)

@siddharthkp
siddharthkp merged commit 47fdbf4 into mainNov 19, 2021
@siddharthkp
siddharthkp deleted the actionlist/disabled-actionlist-item branch November 19, 2021 17:06
This was referenced Nov 19, 2021
@siddharthkpsiddharthkp changed the title ActionList: Visual update for disabled ItemsActionList v2: Visual update for disabled ItemsNov 30, 2021
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@colebemis