Skip to content

Next major - #1639

Merged
jfuchs merged 34 commits into
mainfrom
next_major
Nov 22, 2021
Merged

Next major#1639
jfuchs merged 34 commits into
mainfrom
next_major

Conversation

@jfuchs

Copy link
Copy Markdown
Contributor

Merging the next_major branch into main

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

jfuchsand others added 4 commits November 22, 2021 10:54
* Dialog2 no longer accepts styled system props
* Update docs/content/Dialog2.mdx
Co-authored-by: Mike Perrotti <mperrotti@github.com>
Co-authored-by: Mike Perrotti <mperrotti@github.com>
Co-authored-by: github-actions[bot] <github-actions[bot]@users.noreply.github.com>
@jfuchs
jfuchs requested review from a team, colebemis and jonrohanNovember 22, 2021 21:46
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9d1d84d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/componentsMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js55.76 KB (-0.36% 🔽)
dist/browser.umd.js56.17 KB (-0.37% 🔽)

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Woooo 🚀

@jfuchs
jfuchs merged commit 8ce0eb9 into mainNov 22, 2021
@jfuchs
jfuchs deleted the next_major branch November 22, 2021 21:58
@primer-cssprimer-css mentioned this pull request Nov 22, 2021
matherm-aboehm added a commit to matherm-aboehm/jest-coverage-report-action that referenced this pull request May 13, 2026
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@jfuchs@jonrohan@colebemis@primer-css