Skip to content

Remove "engines" field from package.json - #1712

Merged
colebemis merged 2 commits into
mainfrom
colebemis-patch-2
Dec 9, 2021
Merged

Remove "engines" field from package.json#1712
colebemis merged 2 commits into
mainfrom
colebemis-patch-2

Conversation

@colebemis

@colebemiscolebemis commented Dec 9, 2021

Copy link
Copy Markdown
Contributor

The engines field is breaking builds for projects using node <16. Since node 16 is not required to use @primer/react, let's remove this field from the package.json.

CleanShot 2021-12-09 at 14 53 32@2x

@colebemis
colebemis requested review from a team and rezrahDecember 9, 2021 22:54
@changeset-bot

changeset-botBot commented Dec 9, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2da1e52

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Dec 9, 2021

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js56.98 KB (0%)
dist/browser.umd.js57.29 KB (0%)

@jfuchsjfuchs left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Is engines supposed to be about what's needed to run the distributed package? Or is it about what's needed to build and contribute?

@colebemis

Copy link
Copy Markdown
ContributorAuthor

I think it's "what's needed to run the distributed package" because I couldn't install @primer/react without node 16

@colebemis

Copy link
Copy Markdown
ContributorAuthor

https://docs.npmjs.com/cli/v6/configuring-npm/package-json#engines

You can also use the "engines" field to specify which versions of npm are capable of properly installing your program.

@colebemis
colebemis merged commit 4ce8f07 into mainDec 9, 2021
@colebemis
colebemis deleted the colebemis-patch-2 branch December 9, 2021 23:49
@github-actionsgithub-actionsBot mentioned this pull request Dec 9, 2021
pksjce pushed a commit that referenced this pull request Dec 20, 2021
* Remove "engines" field from package.json
* Create tidy-maps-care.md
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@colebemis@jfuchs