Skip to content

ActionList v2 Divider: Make divider aria-hidden - #1757

Merged
siddharthkp merged 3 commits into
mainfrom
siddharth/actionlist-aria-hidden-divider
Jan 4, 2022
Merged

ActionList v2 Divider: Make divider aria-hidden#1757
siddharthkp merged 3 commits into
mainfrom
siddharth/actionlist-aria-hidden-divider

Conversation

@siddharthkp

Copy link
Copy Markdown
Member

Dividers are decorative, so they should not role="separator" instead they should simply be aria-hidden

Merge checklist

  • Added/updated tests
  • NA Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

@changeset-bot

changeset-botBot commented Dec 22, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 99bfc74

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Dec 22, 2021

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js58.91 KB (0%)
dist/browser.umd.js59.24 KB (0%)

@siddharthkp
siddharthkp merged commit 6e44d78 into mainJan 4, 2022
@siddharthkp
siddharthkp deleted the siddharth/actionlist-aria-hidden-divider branch January 4, 2022 10:52
@primer-cssprimer-css mentioned this pull request Jan 4, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@pksjce