Skip to content

AnchoredOverlay: Add aria-expanded to anchor - #1758

Merged
siddharthkp merged 4 commits into
mainfrom
siddharth/actionmenu2-aria-expanded
Jan 5, 2022
Merged

AnchoredOverlay: Add aria-expanded to anchor#1758
siddharthkp merged 4 commits into
mainfrom
siddharth/actionmenu2-aria-expanded

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Dec 22, 2021

Copy link
Copy Markdown
Member

Updated story + docs for controlled anchor in ActionMenu2

Screenshots

Screen.Recording.2021-12-22.at.3.28.53.PM.mov

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

@siddharthkp
siddharthkp requested review from a team and colebemisDecember 22, 2021 14:32
@changeset-bot

changeset-botBot commented Dec 22, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f562ed9

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkp
siddharthkp requested review from rezrah and removed request for colebemisDecember 22, 2021 14:32
@github-actions

github-actionsBot commented Dec 22, 2021

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js58.91 KB (+0.01% 🔺)
dist/browser.umd.js59.24 KB (+0.01% 🔺)

font-family="normal"
>
<button
aria-expanded="true"

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Wierd that this is a string though

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

yeah!

the types are accepts both, so that's nice

'aria-expanded'?: boolean | 'false' | 'true'; ref: @types/react

@siddharthkp
siddharthkp merged commit b604403 into mainJan 5, 2022
@siddharthkp
siddharthkp deleted the siddharth/actionmenu2-aria-expanded branch January 5, 2022 10:33
@primer-cssprimer-css mentioned this pull request Jan 5, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@pksjce