Skip to content

Export SelectPanel from the package - #1772

Merged
pksjce merged 4 commits into
mainfrom
pk/export-select-panel
Jan 6, 2022
Merged

Export SelectPanel from the package#1772
pksjce merged 4 commits into
mainfrom
pk/export-select-panel

Conversation

@pksjce

Copy link
Copy Markdown
Contributor

As described in #1768, SelectPanel is not exported nor is the documentation displayed.
But the component status page lists it as alpha.
Since it's quite an old component, I'm thinking we missed exporting it.

lmk if its otherwise.

@pksjce
pksjce requested review from a team and rezrahJanuary 5, 2022 07:51
@changeset-bot

changeset-botBot commented Jan 5, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: bc17241

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Jan 5, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js59.92 KB (+1.73% 🔺)
dist/browser.umd.js60.28 KB (+1.76% 🔺)

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Yes please!

@maximedegreve

Copy link
Copy Markdown
Contributor

This works! Just tested it. 🎉

@pksjce
pksjce merged commit 52947c4 into mainJan 6, 2022
@pksjce
pksjce deleted the pk/export-select-panel branch January 6, 2022 03:45
@primer-cssprimer-css mentioned this pull request Jan 6, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@pksjce@maximedegreve@siddharthkp