Skip to content

ActionMenu v2: Use Button v2 as ActionMenu.Button + caret - #1779

Merged
siddharthkp merged 4 commits into
mainfrom
siddharth/actionmenu2-use-button2-with-caret
Jan 19, 2022
Merged

ActionMenu v2: Use Button v2 as ActionMenu.Button + caret#1779
siddharthkp merged 4 commits into
mainfrom
siddharth/actionmenu2-use-button2-with-caret

Conversation

@siddharthkp

Copy link
Copy Markdown
Member

Switched from Button v1 to Button v2 in ActionMenu

@siddharthkp
siddharthkp requested review from a team and pksjceJanuary 7, 2022 10:03
@changeset-bot

changeset-botBot commented Jan 7, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ecaf52f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkp
siddharthkp temporarily deployed to visual-testing January 7, 2022 10:03 Inactive
@siddharthkpsiddharthkp changed the title ActionMenu v2: Use Button v2ActionMenu v2: Use Button v2 as ActionMenu.ButtonJan 7, 2022
@siddharthkpsiddharthkp self-assigned this Jan 7, 2022
@siddharthkp
siddharthkp temporarily deployed to visual-testing January 7, 2022 10:04 Inactive
@siddharthkpsiddharthkp added docs Documentation react labels Jan 7, 2022
@siddharthkp

Copy link
Copy Markdown
MemberAuthor

@pksjce Would love your feedback on docs and stories for usage of Button v2

@github-actions

github-actionsBot commented Jan 7, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js61.25 KB (0%)
dist/browser.umd.js61.63 KB (0%)

@siddharthkpsiddharthkp changed the title ActionMenu v2: Use Button v2 as ActionMenu.ButtonActionMenu v2: Use Button v2 as ActionMenu.Button + caretJan 7, 2022
@siddharthkp
siddharthkp temporarily deployed to visual-testing January 7, 2022 10:44 Inactive
@siddharthkp
siddharthkp merged commit 2630800 into mainJan 19, 2022
@siddharthkp
siddharthkp deleted the siddharth/actionmenu2-use-button2-with-caret branch January 19, 2022 15:33
@primer-cssprimer-css mentioned this pull request Jan 19, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

docsDocumentationreact

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@pksjce